)]}'
{"id":"openstack%2Fglance~668852","triplet_id":"openstack%2Fglance~master~I21e92585f7aa001e317c768cd2b1a9e3c139913a","project":"openstack/glance","branch":"master","topic":"bug/1281453","hashtags":[],"change_id":"I21e92585f7aa001e317c768cd2b1a9e3c139913a","subject":"save_and_reraise_exception instead of \"raise\\n\"","status":"NEW","created":"2019-07-03 09:51:22.000000000","updated":"2019-07-11 13:12:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cd0b0f93cb94ea7acb8f071af10b1fec70b641c","_number":668852,"virtual_id_number":668852,"owner":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-04 08:46:38.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2019-07-04 14:08:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-11 13:12:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2019-07-10 10:04:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 14:37:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 11:48:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-04 08:46:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 14:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 14:37:23.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2019-07-10 10:04:56.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2019-07-11 13:12:42.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"}],"messages":[{"id":"7868de7a2e67a36feab109a52a8e67473c57573a","author":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"date":"2019-07-03 09:51:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a41cf137e77e174f5f89919ae1fcaa797fd4c719","author":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"date":"2019-07-03 10:09:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f9d9fcf2f59e5ee171b8222c347774f35341e53","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-03 11:48:46.000000000","message":"Patch Set 2:\n\nCan you explain why this is wanted instead of just using a built in language mechanism?","accounts_in_message":[],"_revision_number":2},{"id":"f694f2f9a97bc1358e6f4eb4f04042059be44467","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 13:08:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/52/668852/2/check/tempest-full/5672a4e/ : SUCCESS in 1h 59m 59s\n- neutron-grenade http://logs.openstack.org/52/668852/2/check/neutron-grenade/6d7d626/ : SUCCESS in 59m 38s\n- grenade-py3 http://logs.openstack.org/52/668852/2/check/grenade-py3/900cf67/ : SUCCESS in 1h 00m 20s\n- tempest-full-py3 http://logs.openstack.org/52/668852/2/check/tempest-full-py3/962bb61/ : SUCCESS in 1h 41m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/668852/2/check/openstack-tox-lower-constraints/ce14033/ : SUCCESS in 8m 31s\n- openstack-tox-pep8 http://logs.openstack.org/52/668852/2/check/openstack-tox-pep8/0bc314d/ : FAILURE in 6m 36s\n- openstack-tox-py27 http://logs.openstack.org/52/668852/2/check/openstack-tox-py27/75b0353/ : SUCCESS in 9m 52s\n- openstack-tox-py35 http://logs.openstack.org/52/668852/2/check/openstack-tox-py35/2df1561/ : SUCCESS in 8m 04s\n- openstack-tox-py36 http://logs.openstack.org/52/668852/2/check/openstack-tox-py36/4bb5df3/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/52/668852/2/check/openstack-tox-py37/7a16aea/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/52/668852/2/check/openstack-tox-docs/75403cd/ : FAILURE in 5m 54s\n- openstack-tox-functional http://logs.openstack.org/52/668852/2/check/openstack-tox-functional/16ad31d/ : SUCCESS in 9m 20s\n- openstack-tox-functional-py35 http://logs.openstack.org/52/668852/2/check/openstack-tox-functional-py35/3b12e5d/ : SUCCESS in 9m 20s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/52/668852/2/check/glance-eventlet-ssl-handshake-broken-py3/78e63a6/ : FAILURE in 4m 35s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/52/668852/2/check/glance-code-constants-check/fe87b40/ : SUCCESS in 5m 07s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/52/668852/2/check/devstack-plugin-ceph-tempest/11c48c8/ : SUCCESS in 1h 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3119c73b45df9b05fe8d7de80227143879f8e167","author":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"date":"2019-07-03 15:23:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c13d5fa1da9fd5e900e1ec90e16ed268f277430f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 20:11:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/52/668852/3/check/tempest-full/cfb3814/ : SUCCESS in 1h 48m 12s\n- neutron-grenade http://logs.openstack.org/52/668852/3/check/neutron-grenade/2e20dcf/ : SUCCESS in 1h 02m 21s\n- grenade-py3 http://logs.openstack.org/52/668852/3/check/grenade-py3/12a5183/ : SUCCESS in 1h 13m 38s\n- tempest-full-py3 http://logs.openstack.org/52/668852/3/check/tempest-full-py3/a6c0a7b/ : SUCCESS in 1h 46m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/668852/3/check/openstack-tox-lower-constraints/0c2939b/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/52/668852/3/check/openstack-tox-pep8/a303cc0/ : FAILURE in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/52/668852/3/check/openstack-tox-py27/281dfb9/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/52/668852/3/check/openstack-tox-py35/f795e9a/ : SUCCESS in 7m 27s\n- openstack-tox-py36 http://logs.openstack.org/52/668852/3/check/openstack-tox-py36/ec1da42/ : SUCCESS in 6m 37s\n- openstack-tox-py37 http://logs.openstack.org/52/668852/3/check/openstack-tox-py37/73a378e/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/52/668852/3/check/openstack-tox-docs/ba322d8/html/ : SUCCESS in 5m 27s\n- openstack-tox-functional http://logs.openstack.org/52/668852/3/check/openstack-tox-functional/6d8abd7/ : SUCCESS in 10m 25s\n- openstack-tox-functional-py35 http://logs.openstack.org/52/668852/3/check/openstack-tox-functional-py35/2ca2247/ : SUCCESS in 9m 15s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/52/668852/3/check/glance-eventlet-ssl-handshake-broken-py3/b5e34b2/ : FAILURE in 4m 56s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/52/668852/3/check/glance-code-constants-check/ed4d47f/ : SUCCESS in 4m 12s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/52/668852/3/check/devstack-plugin-ceph-tempest/34e8167/ : SUCCESS in 1h 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2b0873aaa25d2618f402cd7e04379208ffb2575c","author":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"date":"2019-07-04 08:45:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"124a787729483a9d86144df999117ee74f85f944","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-04 08:46:38.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"ea370dab665222a93e36cc9a1a80b5f72121d1f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:08:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/52/668852/4/check/tempest-full/631e549/ : SUCCESS in 1h 55m 59s\n- neutron-grenade http://logs.openstack.org/52/668852/4/check/neutron-grenade/4cdd692/ : SUCCESS in 1h 04m 40s\n- grenade-py3 http://logs.openstack.org/52/668852/4/check/grenade-py3/ba5e8a0/ : SUCCESS in 1h 00m 00s\n- tempest-full-py3 http://logs.openstack.org/52/668852/4/check/tempest-full-py3/4eb7945/ : SUCCESS in 1h 32m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/668852/4/check/openstack-tox-lower-constraints/e6167a8/ : SUCCESS in 5m 50s\n- openstack-tox-pep8 http://logs.openstack.org/52/668852/4/check/openstack-tox-pep8/d1681df/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/52/668852/4/check/openstack-tox-py27/aaabca7/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/52/668852/4/check/openstack-tox-py35/556f139/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/52/668852/4/check/openstack-tox-py36/4f0f30e/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/52/668852/4/check/openstack-tox-py37/dc4d58d/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/52/668852/4/check/openstack-tox-docs/9f78f7a/html/ : SUCCESS in 5m 15s\n- openstack-tox-functional http://logs.openstack.org/52/668852/4/check/openstack-tox-functional/c666160/ : SUCCESS in 8m 14s\n- openstack-tox-functional-py35 http://logs.openstack.org/52/668852/4/check/openstack-tox-functional-py35/176dffd/ : SUCCESS in 8m 38s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/52/668852/4/check/glance-eventlet-ssl-handshake-broken-py3/2dee6b0/ : FAILURE in 4m 08s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/52/668852/4/check/glance-code-constants-check/4cecf46/ : SUCCESS in 4m 16s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/52/668852/4/check/devstack-plugin-ceph-tempest/43fcdc9/ : SUCCESS in 1h 43m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d4c3b4643cd36cbbd2e2586605ceaef54d28ffba","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-07-04 14:37:12.000000000","message":"Patch Set 4:\n\n\u003e Can you explain why this is wanted instead of just using a built in\n \u003e language mechanism?\n\nOslo excutils library provides more reach logics for handling exception (e.g. suppress exception given some circumstances, add more logs). Also it prevents (rather unlikely) situation when exception context is cleared and there is attempt to reraise None.\n\nSimilar bug already fixed for neutron some time ago: https://bugs.launchpad.net/neutron/+bug/1279813","accounts_in_message":[],"_revision_number":4},{"id":"2c92936566be3570681388cfb984ff0c8d0b4da6","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-07-04 14:37:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1d7af442f333bf0bd69527d9b57d6d371ffc4d38","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2019-07-04 14:38:23.000000000","message":"Patch Set 4:\n\nPartial fix for the same bug https://review.opendev.org/#/c/78023/","accounts_in_message":[],"_revision_number":4},{"id":"0d3e911ee996f5fc31a4739c45d7419c9204d69d","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2019-07-10 10:04:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a8b65be031115aa2c9e47ad7f5bad53748c87074","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2019-07-11 13:12:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4d3ece8fb30437556f64301353336431e44de966","revisions":{"1250b39b653715d2335f12771df353a211e5f309":{"kind":"REWORK","_number":1,"created":"2019-07-03 09:51:22.000000000","uploader":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"ref":"refs/changes/52/668852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/668852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/668852/1"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 09:10:33.000000000","tz":180},"committer":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 09:51:19.000000000","tz":180},"subject":"save_and_reraise_exception instead of \"raise\\n\"","message":"save_and_reraise_exception instead of \"raise\\n\"\n\nChange-Id: I21e92585f7aa001e317c768cd2b1a9e3c139913a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1250b39b653715d2335f12771df353a211e5f309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1250b39b653715d2335f12771df353a211e5f309"}]},"branch":"refs/heads/master"},"423f30aab5bfa5b5b39c7de32ff92577b51ea44f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-07-03 10:09:44.000000000","uploader":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"ref":"refs/changes/52/668852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/668852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/668852/2"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 09:10:33.000000000","tz":180},"committer":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 10:05:47.000000000","tz":180},"subject":"save_and_reraise_exception instead of \"raise\\n\"","message":"save_and_reraise_exception instead of \"raise\\n\"\n\nChange-Id: I21e92585f7aa001e317c768cd2b1a9e3c139913a\nCloses-bug: 1281453\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/423f30aab5bfa5b5b39c7de32ff92577b51ea44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/423f30aab5bfa5b5b39c7de32ff92577b51ea44f"}]},"branch":"refs/heads/master"},"935bb62ddff9bf1d344ca1637b5b209a4fb12d41":{"kind":"REWORK","_number":3,"created":"2019-07-03 15:23:41.000000000","uploader":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"ref":"refs/changes/52/668852/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/668852/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/668852/3"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 09:10:33.000000000","tz":180},"committer":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 15:22:51.000000000","tz":180},"subject":"save_and_reraise_exception instead of \"raise\\n\"","message":"save_and_reraise_exception instead of \"raise\\n\"\n\nChange-Id: I21e92585f7aa001e317c768cd2b1a9e3c139913a\nCloses-bug: 1281453\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/935bb62ddff9bf1d344ca1637b5b209a4fb12d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/935bb62ddff9bf1d344ca1637b5b209a4fb12d41"}]},"branch":"refs/heads/master"},"4d3ece8fb30437556f64301353336431e44de966":{"kind":"REWORK","_number":4,"created":"2019-07-04 08:45:26.000000000","uploader":{"_account_id":30459,"name":"Artem V. Okhotnikov","email":"okhotnikov.artyom@mail.ru","username":"nominedei"},"ref":"refs/changes/52/668852/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/668852/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/668852/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/668852/4"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-03 09:10:33.000000000","tz":180},"committer":{"name":"Artem Okhotnikov","email":"okhotnikov.artyom@mail.ru","date":"2019-07-04 08:36:42.000000000","tz":180},"subject":"save_and_reraise_exception instead of \"raise\\n\"","message":"save_and_reraise_exception instead of \"raise\\n\"\n\nChange-Id: I21e92585f7aa001e317c768cd2b1a9e3c139913a\nCloses-bug: 1281453\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d3ece8fb30437556f64301353336431e44de966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d3ece8fb30437556f64301353336431e44de966"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
