)]}'
{"id":"openstack%2Fglance~669655","triplet_id":"openstack%2Fglance~master~I591e2637bdef9c7bf01b021b3eec2a81aa45c21d","project":"openstack/glance","branch":"master","topic":"bp/cluster-awareness","hashtags":[],"change_id":"I591e2637bdef9c7bf01b021b3eec2a81aa45c21d","subject":"Cluster aware Glance with RabbitMQ","status":"ABANDONED","created":"2019-07-08 13:00:15.000000000","updated":"2021-03-03 06:19:43.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"46ea21ccc0486988ea9691122a6ee58e45f6a059","_number":669655,"virtual_id_number":669655,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-06 05:04:16.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"date":"2019-09-05 04:55:43.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2019-08-30 12:16:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-28 20:34:58.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"date":"2019-09-02 02:58:04.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 16:08:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-08-28 20:34:58.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-08-30 12:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-02 02:58:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2019-09-05 04:55:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-09-06 05:04:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f25c6203353d6f555859053bac0661f9629b54c3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 13:00:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c35d8a5125ec706290544d160c5a9e00e7f941a2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-08 13:00:37.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"23915ba99db07b4da7f757b58cebf00986009d5a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-07-08 16:08:30.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nI\u0027m not a glance expert (of course), but these are just a few comments from a layperson from reading the code :)","accounts_in_message":[],"_revision_number":1},{"id":"d08a07a4bbfc600bea8dd80970b589b946ba9098","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 19:12:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/55/669655/1/check/requirements-check/80ab806/ : FAILURE in 3m 15s\n- tempest-full http://logs.openstack.org/55/669655/1/check/tempest-full/219de0b/ : SUCCESS in 1h 38m 03s\n- neutron-grenade http://logs.openstack.org/55/669655/1/check/neutron-grenade/8286e94/ : SUCCESS in 1h 01m 46s\n- grenade-py3 http://logs.openstack.org/55/669655/1/check/grenade-py3/4c3f155/ : SUCCESS in 1h 03m 14s\n- tempest-full-py3 http://logs.openstack.org/55/669655/1/check/tempest-full-py3/338f22a/ : SUCCESS in 1h 31m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/669655/1/check/openstack-tox-lower-constraints/46628b4/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/55/669655/1/check/openstack-tox-pep8/88161c5/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/55/669655/1/check/openstack-tox-py27/172ddb0/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/55/669655/1/check/openstack-tox-py35/0a99baf/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/55/669655/1/check/openstack-tox-py36/5db60f9/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/55/669655/1/check/openstack-tox-py37/3ae8057/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/55/669655/1/check/openstack-tox-docs/a7bb6ce/html/ : SUCCESS in 4m 50s\n- openstack-tox-functional http://logs.openstack.org/55/669655/1/check/openstack-tox-functional/758c3b7/ : SUCCESS in 10m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/55/669655/1/check/openstack-tox-functional-py35/9bb4f1f/ : SUCCESS in 8m 38s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/55/669655/1/check/glance-eventlet-ssl-handshake-broken-py3/f87c8ea/ : FAILURE in 4m 18s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/55/669655/1/check/glance-code-constants-check/f3b0b46/ : SUCCESS in 4m 30s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/55/669655/1/check/devstack-plugin-ceph-tempest/e587029/ : SUCCESS in 1h 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e1daab1d6a5deffa8cb1a53dfdd54dcc0c13693","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 20:00:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0053cded6974049bad19238b2c342779939c4c7c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-08 20:01:22.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"e1aab11365d205cf3d272e347f934cc749b57eaf","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-08 20:02:45.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nThanks, Dan","accounts_in_message":[],"_revision_number":1},{"id":"681791f27263e198748b4df881629fd9947d1f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 01:09:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/55/669655/2/check/requirements-check/f78fb51/ : FAILURE in 3m 36s\n- tempest-full http://logs.openstack.org/55/669655/2/check/tempest-full/662b271/ : SUCCESS in 1h 45m 17s\n- neutron-grenade http://logs.openstack.org/55/669655/2/check/neutron-grenade/fc5c382/ : SUCCESS in 1h 01m 11s\n- grenade-py3 http://logs.openstack.org/55/669655/2/check/grenade-py3/416619f/ : SUCCESS in 1h 03m 13s\n- tempest-full-py3 http://logs.openstack.org/55/669655/2/check/tempest-full-py3/b96713c/ : SUCCESS in 2h 03m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/669655/2/check/openstack-tox-lower-constraints/977014c/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/55/669655/2/check/openstack-tox-pep8/43b312b/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/55/669655/2/check/openstack-tox-py27/7a2bcb2/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/55/669655/2/check/openstack-tox-py35/50364d9/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/55/669655/2/check/openstack-tox-py36/3c6fcc5/ : SUCCESS in 5m 15s\n- openstack-tox-py37 http://logs.openstack.org/55/669655/2/check/openstack-tox-py37/5fd4dde/ : SUCCESS in 6m 59s\n- openstack-tox-docs http://logs.openstack.org/55/669655/2/check/openstack-tox-docs/93dba0a/html/ : SUCCESS in 5m 31s\n- openstack-tox-functional http://logs.openstack.org/55/669655/2/check/openstack-tox-functional/c140341/ : SUCCESS in 9m 36s\n- openstack-tox-functional-py35 http://logs.openstack.org/55/669655/2/check/openstack-tox-functional-py35/448b6fb/ : FAILURE in 11m 02s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/55/669655/2/check/glance-eventlet-ssl-handshake-broken-py3/fbafe8d/ : FAILURE in 4m 11s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/55/669655/2/check/glance-code-constants-check/ec9945c/ : SUCCESS in 3m 51s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/55/669655/2/check/devstack-plugin-ceph-tempest/d1b42f8/ : SUCCESS in 1h 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b940a0a2d13735b0a298b44c2a9a965106c03779","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-09 08:02:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"179c87cb720eaa3257aab9bda378bda190226272","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-09 08:03:26.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"9a99dfb156106be01211b5ec59942e3ce0a98356","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-09 10:56:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nWorking on single node, need to check on multiple controller nodes.","accounts_in_message":[],"_revision_number":3},{"id":"05f4d1c79d0078e8ee3da5df774959b54a711841","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-07-09 11:08:53.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3dc8c65ea158903fd115125d79d0f33d68715b51","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-09 11:09:12.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"909b4743da6efc8ce1524c4afec818c52a44ca96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:02:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/55/669655/4/check/requirements-check/fdc3453/ : SUCCESS in 2m 22s\n- tempest-full http://logs.openstack.org/55/669655/4/check/tempest-full/bc62861/ : SUCCESS in 2h 00m 59s\n- neutron-grenade http://logs.openstack.org/55/669655/4/check/neutron-grenade/106fb81/ : SUCCESS in 1h 16m 21s\n- grenade-py3 http://logs.openstack.org/55/669655/4/check/grenade-py3/cde8c93/ : SUCCESS in 1h 06m 12s\n- tempest-full-py3 http://logs.openstack.org/55/669655/4/check/tempest-full-py3/ca9f775/ : SUCCESS in 1h 34m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/669655/4/check/openstack-tox-lower-constraints/97ee64c/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/55/669655/4/check/openstack-tox-pep8/37a597d/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/55/669655/4/check/openstack-tox-py27/594ede4/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/55/669655/4/check/openstack-tox-py35/0712c4e/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/55/669655/4/check/openstack-tox-py36/16cf4be/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/55/669655/4/check/openstack-tox-py37/f23fb61/ : FAILURE in 7m 21s\n- openstack-tox-docs http://logs.openstack.org/55/669655/4/check/openstack-tox-docs/d898f7c/html/ : SUCCESS in 5m 19s\n- openstack-tox-functional http://logs.openstack.org/55/669655/4/check/openstack-tox-functional/ddac88b/ : SUCCESS in 8m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/55/669655/4/check/openstack-tox-functional-py35/b7c5694/ : SUCCESS in 9m 29s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/55/669655/4/check/glance-eventlet-ssl-handshake-broken-py3/883f8ca/ : FAILURE in 4m 10s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/55/669655/4/check/glance-code-constants-check/ca22f6d/ : SUCCESS in 4m 35s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/55/669655/4/check/devstack-plugin-ceph-tempest/2e42d1b/ : SUCCESS in 1h 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a1c92dcd9b429da945b2052f9e96783de933bdb1","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-07 14:25:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8b609266b131370030e907f48b31a93b725cca0e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-07 15:10:12.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/55/669655/4/check-not-vote/ext-glance-zuul/82e14cc : FAILURE in 1m 30s","accounts_in_message":[],"_revision_number":4},{"id":"dd8ed026c05b47024fd69e0da8cade8947054d87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:36:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/55/669655/4/check/requirements-check/de8c0fd/ : SUCCESS in 3m 28s\n- tempest-full https://logs.opendev.org/55/669655/4/check/tempest-full/7bb21a0/ : SUCCESS in 1h 48m 13s\n- neutron-grenade https://logs.opendev.org/55/669655/4/check/neutron-grenade/1237565/ : SUCCESS in 1h 02m 09s\n- grenade-py3 https://logs.opendev.org/55/669655/4/check/grenade-py3/c3514db/ : SUCCESS in 1h 10m 30s\n- tempest-full-py3 https://logs.opendev.org/55/669655/4/check/tempest-full-py3/cb4e881/ : SUCCESS in 1h 24m 43s\n- openstack-tox-lower-constraints https://logs.opendev.org/55/669655/4/check/openstack-tox-lower-constraints/dfb7b49/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://logs.opendev.org/55/669655/4/check/openstack-tox-pep8/d7b7d89/ : SUCCESS in 5m 57s\n- openstack-tox-py27 https://logs.opendev.org/55/669655/4/check/openstack-tox-py27/7981a6d/ : SUCCESS in 5m 04s\n- openstack-tox-py35 https://logs.opendev.org/55/669655/4/check/openstack-tox-py35/bfad76a/ : SUCCESS in 7m 20s\n- openstack-tox-py36 https://logs.opendev.org/55/669655/4/check/openstack-tox-py36/ce600ce/ : SUCCESS in 5m 55s\n- openstack-tox-py37 https://logs.opendev.org/55/669655/4/check/openstack-tox-py37/7debfe4/ : SUCCESS in 6m 35s\n- openstack-tox-docs https://logs.opendev.org/55/669655/4/check/openstack-tox-docs/2435ef9/html/ : SUCCESS in 6m 03s\n- openstack-tox-functional https://logs.opendev.org/55/669655/4/check/openstack-tox-functional/95bfc36/ : SUCCESS in 16m 55s\n- openstack-tox-functional-py35 https://logs.opendev.org/55/669655/4/check/openstack-tox-functional-py35/2441103/ : FAILURE in 13m 14s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/55/669655/4/check/glance-eventlet-ssl-handshake-broken-py3/29b7a6f/ : FAILURE in 6m 51s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/55/669655/4/check/glance-code-constants-check/6c49837/ : SUCCESS in 5m 44s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/55/669655/4/check/devstack-plugin-ceph-tempest/09a3d6c/ : SUCCESS in 1h 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"006927c79e6ef734099163b6d4112388ff3d7bde","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-08 12:59:04.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"19f691f9f294c01a15410ba687e3e142dd0ff319","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 18:03:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/55/669655/5/check/requirements-check/2a02430/ : SUCCESS in 3m 34s\n- tempest-full https://logs.opendev.org/55/669655/5/check/tempest-full/e75e002/ : SUCCESS in 1h 39m 08s\n- neutron-grenade https://logs.opendev.org/55/669655/5/check/neutron-grenade/4a6a9a3/ : SUCCESS in 58m 25s\n- grenade-py3 https://logs.opendev.org/55/669655/5/check/grenade-py3/e301d75/ : SUCCESS in 1h 00m 31s\n- tempest-full-py3 https://logs.opendev.org/55/669655/5/check/tempest-full-py3/30ad131/ : SUCCESS in 1h 26m 32s\n- openstack-tox-lower-constraints https://logs.opendev.org/55/669655/5/check/openstack-tox-lower-constraints/1c7e9be/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://logs.opendev.org/55/669655/5/check/openstack-tox-pep8/794b806/ : SUCCESS in 5m 33s\n- openstack-tox-py27 https://logs.opendev.org/55/669655/5/check/openstack-tox-py27/79b073e/ : SUCCESS in 4m 07s\n- openstack-tox-py35 https://logs.opendev.org/55/669655/5/check/openstack-tox-py35/45e242a/ : SUCCESS in 7m 18s\n- openstack-tox-py36 https://logs.opendev.org/55/669655/5/check/openstack-tox-py36/03d0de0/ : FAILURE in 6m 31s\n- openstack-tox-py37 https://logs.opendev.org/55/669655/5/check/openstack-tox-py37/494df52/ : SUCCESS in 7m 33s\n- openstack-tox-docs https://logs.opendev.org/55/669655/5/check/openstack-tox-docs/2b0a8d5/html/ : SUCCESS in 4m 54s\n- openstack-tox-functional https://logs.opendev.org/55/669655/5/check/openstack-tox-functional/874b755/ : SUCCESS in 9m 22s\n- openstack-tox-functional-py35 https://logs.opendev.org/55/669655/5/check/openstack-tox-functional-py35/8044935/ : SUCCESS in 9m 35s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/55/669655/5/check/glance-eventlet-ssl-handshake-broken-py3/4ec9bcb/ : FAILURE in 5m 37s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/55/669655/5/check/glance-code-constants-check/c4766c1/ : SUCCESS in 5m 46s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/55/669655/5/check/devstack-plugin-ceph-tempest/7c92391/ : FAILURE in 1h 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"072c9169614d9984554c185e883ccaef71fb71eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-09 05:30:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ab8030d158de3ffad4ab7bc8c292bf04ebc4e6a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 07:12:18.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/55/669655/5/check/requirements-check/9c4aced/ : SUCCESS in 4m 02s\n- tempest-full https://logs.opendev.org/55/669655/5/check/tempest-full/a21d2df/ : SUCCESS in 1h 36m 31s\n- neutron-grenade https://logs.opendev.org/55/669655/5/check/neutron-grenade/b1416f5/ : FAILURE in 52m 13s\n- grenade-py3 https://logs.opendev.org/55/669655/5/check/grenade-py3/68fe7af/ : SUCCESS in 56m 48s\n- tempest-full-py3 https://logs.opendev.org/55/669655/5/check/tempest-full-py3/82e1d7e/ : SUCCESS in 1h 18m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/55/669655/5/check/openstack-tox-lower-constraints/cc496f8/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://logs.opendev.org/55/669655/5/check/openstack-tox-pep8/55c2cb9/ : SUCCESS in 4m 32s\n- openstack-tox-py27 https://logs.opendev.org/55/669655/5/check/openstack-tox-py27/b7f889a/ : SUCCESS in 4m 19s\n- openstack-tox-py35 https://logs.opendev.org/55/669655/5/check/openstack-tox-py35/42197b1/ : SUCCESS in 5m 13s\n- openstack-tox-py36 https://logs.opendev.org/55/669655/5/check/openstack-tox-py36/d5dabfa/ : SUCCESS in 4m 25s\n- openstack-tox-py37 https://logs.opendev.org/55/669655/5/check/openstack-tox-py37/5e53111/ : SUCCESS in 5m 04s\n- openstack-tox-docs https://logs.opendev.org/55/669655/5/check/openstack-tox-docs/4dcdbbb/html/ : SUCCESS in 4m 50s\n- openstack-tox-functional https://logs.opendev.org/55/669655/5/check/openstack-tox-functional/83bfe6e/ : SUCCESS in 8m 00s\n- openstack-tox-functional-py35 https://logs.opendev.org/55/669655/5/check/openstack-tox-functional-py35/7c765c0/ : SUCCESS in 9m 38s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/55/669655/5/check/glance-eventlet-ssl-handshake-broken-py3/b456fa3/ : FAILURE in 4m 20s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/55/669655/5/check/glance-code-constants-check/bc1fb46/ : SUCCESS in 4m 17s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/55/669655/5/check/devstack-plugin-ceph-tempest/39dcdfa/ : SUCCESS in 1h 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e9ab65e6f05bdc8112058bb6868fabb249c2f505","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-09 07:16:40.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ba2680bf4d7b3e66abecd7253dd720e1e34b22dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 10:00:41.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://logs.opendev.org/55/669655/5/check/requirements-check/1c9ccd1/ : SUCCESS in 4m 04s\n- tempest-full https://logs.opendev.org/55/669655/5/check/tempest-full/f090228/ : SUCCESS in 1h 38m 11s\n- neutron-grenade https://logs.opendev.org/55/669655/5/check/neutron-grenade/c7b57b1/ : FAILURE in 51m 15s\n- grenade-py3 https://logs.opendev.org/55/669655/5/check/grenade-py3/0bc94ec/ : SUCCESS in 1h 07m 40s\n- tempest-full-py3 https://logs.opendev.org/55/669655/5/check/tempest-full-py3/efbc173/ : SUCCESS in 1h 47m 04s\n- openstack-tox-lower-constraints https://logs.opendev.org/55/669655/5/check/openstack-tox-lower-constraints/1a0277f/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://logs.opendev.org/55/669655/5/check/openstack-tox-pep8/15d39ba/ : SUCCESS in 4m 18s\n- openstack-tox-py27 https://logs.opendev.org/55/669655/5/check/openstack-tox-py27/1ad8eef/ : SUCCESS in 4m 44s\n- openstack-tox-py35 https://logs.opendev.org/55/669655/5/check/openstack-tox-py35/4b4a360/ : SUCCESS in 5m 51s\n- openstack-tox-py36 https://logs.opendev.org/55/669655/5/check/openstack-tox-py36/2ac078d/ : SUCCESS in 5m 16s\n- openstack-tox-py37 https://logs.opendev.org/55/669655/5/check/openstack-tox-py37/b5d7970/ : SUCCESS in 6m 45s\n- openstack-tox-docs https://logs.opendev.org/55/669655/5/check/openstack-tox-docs/c3da083/html/ : SUCCESS in 5m 32s\n- openstack-tox-functional https://logs.opendev.org/55/669655/5/check/openstack-tox-functional/16da8ac/ : SUCCESS in 9m 12s\n- openstack-tox-functional-py35 https://logs.opendev.org/55/669655/5/check/openstack-tox-functional-py35/de4ef34/ : SUCCESS in 10m 37s\n- glance-eventlet-ssl-handshake-broken-py3 https://logs.opendev.org/55/669655/5/check/glance-eventlet-ssl-handshake-broken-py3/e013699/ : FAILURE in 4m 35s (non-voting)\n- glance-code-constants-check https://logs.opendev.org/55/669655/5/check/glance-code-constants-check/2e44e48/ : SUCCESS in 5m 32s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/55/669655/5/check/devstack-plugin-ceph-tempest/8171e40/ : SUCCESS in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e011e0de32792112578a0bb22430305bb6c353f4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-09 10:16:41.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/55/669655/5/check-not-vote/ext-glance-zuul/a00a88a : FAILURE in 16m 15s","accounts_in_message":[],"_revision_number":5},{"id":"ddfeb7e8a3e93eda372e6da2c0dfdaecbd54ef81","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-08-28 20:34:58.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"95e6190e7a46ee6264d7f18b2912539b85650724","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 22:19:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/73e3adadc1ad49de8e2ddd35fa9977ea : SUCCESS in 2m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/650f5e364c36490ea6b7fb3d0faa3b3b : SUCCESS in 1h 33m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/55204f4d3043486b933027de459fa1c5 : SUCCESS in 56m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/48e326e62a5649adb180fc896134225b : SUCCESS in 1h 17m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d83206fa9824d8fbfbb7ce38ba92be6 : SUCCESS in 1h 37m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73035830688f4fafa9021bd920337aaf : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b8e7218d664760aec05db40cd7cbff : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/838532a3751e48cd967f9398faade46b : SUCCESS in 5m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bbdcd46a66ca4b928829a684bbe7a313 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e844659ca38a4147abe72756c9a5dddc : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/351ae72c23ad4a3d9f44d29f75d698ae : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0cf9f33d2b49ce990f4e5799329cd5 : SUCCESS in 6m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1d78068b6e2d4a6e9e95ee53165f96a1 : SUCCESS in 8m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/68519ae50031469fa5ab137fc31ea8ba : SUCCESS in 9m 27s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/92e6afa5b8e4487e85af2efa7ea9d8d2 : FAILURE in 5m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/73ab0c01479140418929f081609ce79a : SUCCESS in 5m 02s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd67c65ec4d045a3a21610e0b2f47675 : SUCCESS in 1h 09m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"007b45f0155a297ae571e8a4b19975f9d6f8d3a9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 06:32:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"79991c1db62df39372775b948d8cf296ef7a6bc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:50:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ef335edac784cdf836b797329722a8b : SUCCESS in 3m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6272b62aae384a4fa8a49fb344cc8140 : FAILURE in 1h 28m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4dff5155608c4f32bc29bee4a662c2dc : SUCCESS in 1h 00m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d688e8b671914d85b1e0dd1aed20d649 : SUCCESS in 1h 09m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/330c74d123b34e4991bf7af5587911d3 : SUCCESS in 1h 27m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1c92704cef44dfa9202ffac51e0194a : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c3be4a6aeb41ee93045916e8157ed9 : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cd4fa32aba84d8fab36ec4e26fecc2b : SUCCESS in 6m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/429a241f44ee43eb8f504109e83633cc : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a2d4ae813a641a2b084f1239408e98a : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7c5a99741854531aebbb183b6f1fdad : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9afdef1e7d3a4ed392de99d9f9fa031d : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0f0936d5d4504915946420d1250c80bb : SUCCESS in 9m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5c640b92354344ccae5a1b5e8e98bd3f : FAILURE in 16m 16s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/9907893865a0453d98cd114fa5e36afa : FAILURE in 9m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6320443b75684ebbbc0f8fe141dd0b89 : SUCCESS in 7m 55s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/03c4d51e4cef4c69ad1a5258e418a12e : SUCCESS in 1h 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f6b9e2e5193c77e98142c0a36274865c5919bb04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-30 09:21:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1ec01bfbb3b031d53cac32472c6c570e490619f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 12:16:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4886f5491989450ba12f1a63fbeeff50 : SUCCESS in 3m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7f96a73bc0704801974d61b8757d18ad : SUCCESS in 1h 53m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d61585fc8a345bf8a2c296116fdea95 : SUCCESS in 1h 03m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee5b722ee4f74d6b96905832f21378d1 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a865f464e7b4348a16e5591af3d10c9 : SUCCESS in 1h 28m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c114f4c9c91940208831b779fe6a5eb0 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c37318f2dbc41f58c1e5eac675986e7 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5697d924969f4e01a320060edb870c66 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/51b1f6ce88d643beafed5e820afc6456 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d7b4bcbf4d4f728dedc4e80a47124b : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1e1cb380159422ba684b2df6ab47bf6 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9afd634b4e42c9b1a664ba273a97bf : SUCCESS in 9m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/23fc020170144f40b74442d92c808f5b : SUCCESS in 11m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6b8a99da71ec4cfd9160b01d04d5136e : SUCCESS in 9m 43s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/f21d154206e045d89c4db25ed99d0ff0 : FAILURE in 5m 29s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7d05763e56e64f3594327367fdbf16e4 : SUCCESS in 5m 29s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e8451b48e934c5a9c5585dee3b2e79e : SUCCESS in 1h 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a58a537cd6aeb0e6f5a6ccde98abcc51a0bed02","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-09-02 02:58:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fd78d4f3a62903b3ad876c0409fd042ee25a4725","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-04 05:56:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"94a45f6bca37b9eabe8d94d212d1c077e32324f7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-04 05:56:49.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"9ae6ffd3a0b7ee6ea6c8e815defb7864b8451dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 07:51:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7d1d370c6db491294ad28378cd8159a : SUCCESS in 3m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5e6d8926e069413e938b7ae8504e49b2 : SUCCESS in 1h 47m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a2e84558b8d94e448c188b3d73f745a7 : SUCCESS in 1h 00m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a80c436e13924ef88fa2cbc658a2437a : SUCCESS in 1h 00m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8974c39d6a847cbaee50bb36d3503cc : SUCCESS in 1h 15m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b862458f9b4c49cd9a8d0effe4ba4ee3 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab6f293add144db99a41420a9f60a3f : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da80cd4c4e92444184b906a2cf268898 : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0b1783c5a02f492384826140f393d8e2 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2800819fc2a48e38ff29d43fc197a9e : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/672bdb6902cd4bc09df06c09ed739759 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd25b5a88fa4a1bad265e332b372213 : SUCCESS in 9m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d9d9d39d501e4551afff35d351d5ff88 : SUCCESS in 9m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e4eae0b152734225ad21c45225e0e2b7 : SUCCESS in 9m 59s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/b812e9bd9f95457397c3da8410533ab4 : FAILURE in 6m 44s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/02dd3d30f01a4c47b7d252e94011c1d9 : SUCCESS in 5m 56s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c71d7d7c7c048bc8c4a0e786260361c : FAILURE in 1h 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2ea2666bb02ec7df9b453bc2fe22f0b036b2dd70","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-04 07:53:27.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"561966eb02fe9b33adedbc8d24c8ec816434b3f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 09:56:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0357c80ac803431b88056add2b4352d3 : SUCCESS in 5m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9a0db4fd3dff42d1beaeefea0f8fd95c : SUCCESS in 2h 00m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0d13d0ebace74b3db4875c6ccb595e2f : SUCCESS in 1h 22m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2e67a1cc376247f8b339fd323423ac4a : SUCCESS in 1h 23m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07ee4218bda44050b720cafa4fd81a66 : SUCCESS in 1h 45m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62359b3a7d9d4114ad99b22b276e0b3b : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/444f643167074973ac08ea38817026a8 : SUCCESS in 8m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8a645c5808740ae93efa820d1a5b194 : SUCCESS in 9m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4ff9d613e1c7465fb4dc5f5a072fba95 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c30b0748957f49bd904ba044a83e7302 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fc01ade903542eb9ca344e2cb05897c : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d65b11f649a4de5ad6073b82c8fa998 : SUCCESS in 8m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6590395314064a13a62c86a141678ada : SUCCESS in 9m 20s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/758e6b367aef48f5be11b5b466b02ca4 : SUCCESS in 9m 38s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/d3a1fc349f594d989199e40566a0dfc7 : FAILURE in 4m 53s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/61c974d94144451983ff686f586d7214 : SUCCESS in 3m 50s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/752957d0b31047919f12ebb1bfdf7981 : FAILURE in 1h 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b11ec722eca6fd710ac0027ace0e610427a2033","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-04 19:16:22.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/55/669655/5/check-not-vote/ext-glance-zuul/51d5edf : FAILURE in 42m 59s","accounts_in_message":[],"_revision_number":5},{"id":"d0c9f575ab942fd19412678eda600d6cd66e566f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-05 04:55:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"974861d175a997cd6576ecdbd0976bb96c6efa7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 00:42:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c69ddc9bd954e3c91f8529b0e808fbd : SUCCESS in 4m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9b4746788c9748e68d969e0411b9720a : SUCCESS in 1h 35m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8edb5a77edb7427392c6dfb530d138aa : SUCCESS in 1h 07m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0370e7e592d846729f15430dc211e1bd : SUCCESS in 1h 13m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df32e0062fd14d488c32c263a7b449d4 : SUCCESS in 1h 28m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe9f881ba36040b99010295f6272040a : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345bf113f73d43dd853d98346779c454 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/493c58d35ee049a2ba8ffce3a738c9c5 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/414a4a8c8a194d928fa6b6be3fc2ccbb : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ac5d8acbde442a39a989b6800f6d9ef : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24489abaa614437a1a078bc3f4e96e4 : SUCCESS in 8m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b31f2e3e7aa849beb92e5ef3dfe96548 : SUCCESS in 9m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c0ed49c3e2ab4c959129cbd917b65a3a : SUCCESS in 8m 13s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/b118bfd04da14ee8856dd29e6d4c86cd : FAILURE in 4m 37s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/63ecc2caf0174054af2c26c4678426e8 : SUCCESS in 6m 40s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e18a717a4a264aa2b220b5caf66e9aed : SUCCESS in 1h 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa1d4428d2343396c2216aa5b40439833990b399","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-06 05:04:16.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/55/669655/5/check-not-vote/ext-glance-zuul/93f1b45 : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":5},{"id":"46ea21ccc0486988ea9691122a6ee58e45f6a059","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-03 06:19:43.000000000","message":"Abandoned\n\nFeel free to restore once we finalize the implementation design","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5053386488339e7c9552309361e8b6a14d5777fb","revisions":{"40ab7842a17c404314b532696660deb36f1ff99f":{"kind":"REWORK","_number":1,"created":"2019-07-08 13:00:15.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/55/669655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/669655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/669655/1"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 05:58:02.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 12:59:12.000000000","tz":60},"subject":"Cluster aware Glance with RabbitMQ","message":"Cluster aware Glance with RabbitMQ\n\nThis change provides the basic functionality of cluster\nmessaging for glance-api services.\n\nInitial usecase the \u0027glance-direct\u0027 image import is\naddressed where the nodes can redistribute the import\nrequest to other glance-api nodes for processing avoiding\nthe need of shared filesystem between the nodes for staging.\n\nAs rabbitMQ requires x509 auth for TLS communication it\u0027s\nout of the scope of this change and should be addressed on\nfollow-up patch.\n\nSuitable test cases should be added once identified.\n\nrelated-bp: cluster-awareness\n\nChange-Id: I591e2637bdef9c7bf01b021b3eec2a81aa45c21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40ab7842a17c404314b532696660deb36f1ff99f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40ab7842a17c404314b532696660deb36f1ff99f"}]},"branch":"refs/heads/master"},"fb4a581828e16c2795c839ae97234c19fb23f78a":{"kind":"REWORK","_number":2,"created":"2019-07-08 20:00:52.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/55/669655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/669655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/669655/2"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 05:58:02.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 20:00:36.000000000","tz":60},"subject":"Cluster aware Glance with RabbitMQ","message":"Cluster aware Glance with RabbitMQ\n\nThis change provides the basic functionality of cluster\nmessaging for glance-api services.\n\nInitial usecase the \u0027glance-direct\u0027 image import is\naddressed where the nodes can redistribute the import\nrequest to other glance-api nodes for processing avoiding\nthe need of shared filesystem between the nodes for staging.\n\nAs rabbitMQ requires x509 auth for TLS communication it\u0027s\nout of the scope of this change and should be addressed on\nfollow-up patch.\n\nSuitable test cases should be added once identified.\n\nrelated-bp: cluster-awareness\n\nChange-Id: I591e2637bdef9c7bf01b021b3eec2a81aa45c21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb4a581828e16c2795c839ae97234c19fb23f78a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb4a581828e16c2795c839ae97234c19fb23f78a"}]},"branch":"refs/heads/master"},"bd2be7c691d5de06b36c467dd6b26ac417586216":{"kind":"REWORK","_number":3,"created":"2019-07-09 08:02:58.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/55/669655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/669655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/669655/3"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 05:58:02.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-09 08:02:38.000000000","tz":60},"subject":"Cluster aware Glance with RabbitMQ","message":"Cluster aware Glance with RabbitMQ\n\nThis change provides the basic functionality of cluster\nmessaging for glance-api services.\n\nInitial usecase the \u0027glance-direct\u0027 image import is\naddressed where the nodes can redistribute the import\nrequest to other glance-api nodes for processing avoiding\nthe need of shared filesystem between the nodes for staging.\n\nAs rabbitMQ requires x509 auth for TLS communication it\u0027s\nout of the scope of this change and should be addressed on\nfollow-up patch.\n\nSuitable test cases should be added once identified.\n\nrelated-bp: cluster-awareness\n\nChange-Id: I591e2637bdef9c7bf01b021b3eec2a81aa45c21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd2be7c691d5de06b36c467dd6b26ac417586216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd2be7c691d5de06b36c467dd6b26ac417586216"}]},"branch":"refs/heads/master"},"0336ed3c88eaece3e1f11bf0c8a2f881ba304fe4":{"kind":"REWORK","_number":4,"created":"2019-07-09 11:08:53.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/55/669655/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/669655/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/669655/4"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 05:58:02.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-09 11:08:48.000000000","tz":0},"subject":"Cluster aware Glance with RabbitMQ","message":"Cluster aware Glance with RabbitMQ\n\nThis change provides the basic functionality of cluster\nmessaging for glance-api services.\n\nInitial usecase the \u0027glance-direct\u0027 image import is\naddressed where the nodes can redistribute the import\nrequest to other glance-api nodes for processing avoiding\nthe need of shared filesystem between the nodes for staging.\n\nAs rabbitMQ requires x509 auth for TLS communication it\u0027s\nout of the scope of this change and should be addressed on\nfollow-up patch.\n\nSuitable test cases should be added once identified.\n\nrelated-bp: cluster-awareness\n\nChange-Id: I591e2637bdef9c7bf01b021b3eec2a81aa45c21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0336ed3c88eaece3e1f11bf0c8a2f881ba304fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0336ed3c88eaece3e1f11bf0c8a2f881ba304fe4"}]},"branch":"refs/heads/master"},"5053386488339e7c9552309361e8b6a14d5777fb":{"kind":"REWORK","_number":5,"created":"2019-08-08 12:59:04.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/55/669655/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/669655/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/669655/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/669655/5"}}},"commit":{"parents":[{"commit":"90b1f8dc7a2411a7e0286f286d84d8e511d93791","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90b1f8dc7a2411a7e0286f286d84d8e511d93791"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-07-08 05:58:02.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-08-08 12:59:00.000000000","tz":0},"subject":"Cluster aware Glance with RabbitMQ","message":"Cluster aware Glance with RabbitMQ\n\nThis change provides the basic functionality of cluster\nmessaging for glance-api services.\n\nInitial usecase the \u0027glance-direct\u0027 image import is\naddressed where the nodes can redistribute the import\nrequest to other glance-api nodes for processing avoiding\nthe need of shared filesystem between the nodes for staging.\n\nAs rabbitMQ requires x509 auth for TLS communication it\u0027s\nout of the scope of this change and should be addressed on\nfollow-up patch.\n\nSuitable test cases should be added once identified.\n\nrelated-bp: cluster-awareness\n\nChange-Id: I591e2637bdef9c7bf01b021b3eec2a81aa45c21d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5053386488339e7c9552309361e8b6a14d5777fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5053386488339e7c9552309361e8b6a14d5777fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
