)]}'
{"id":"openstack%2Fglance~670746","triplet_id":"openstack%2Fglance~master~I28b01dc028427bc08609d2f30031bda547a4c075","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I28b01dc028427bc08609d2f30031bda547a4c075","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2019-07-15 01:12:22.000000000","updated":"2019-07-15 02:12:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3c89b9cf617e155da7f90c8e1b75cc32d14ca5a","_number":670746,"virtual_id_number":670746,"owner":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-15 01:16:01.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-15 01:25:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 01:16:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-15 01:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f60e100bd4c225da70cb81a49dbb11390f6b43a6","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-07-15 01:12:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c972da9313a7fcdddceab904ef0f885fbf5ee139","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-15 01:12:41.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"c6f8eb76350e130dc941502c31ce986838d1d82c","author":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"date":"2019-07-15 01:15:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"89c2b27930d4e894a72286e3cc3b6653027724a2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-15 01:16:01.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"2514db3b67a5fbcbe6a166e4e40e94b15367015d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 01:25:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/46/670746/2/check/test-release-openstack/8cd02d3/ : SUCCESS in 2m 38s\n- build-openstack-api-ref http://logs.openstack.org/46/670746/2/check/build-openstack-api-ref/31da78d/html/ : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/46/670746/2/check/openstack-tox-lower-constraints/a1f6ce2/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/46/670746/2/check/openstack-tox-pep8/ee79dff/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/46/670746/2/check/openstack-tox-py27/050a7ac/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/46/670746/2/check/openstack-tox-py36/43b6128/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/46/670746/2/check/openstack-tox-py37/584093c/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/46/670746/2/check/openstack-tox-docs/1ef7348/html/ : SUCCESS in 4m 32s\n- build-openstack-releasenotes http://logs.openstack.org/46/670746/2/check/build-openstack-releasenotes/e4572c2/html/ : SUCCESS in 8m 06s\n- openstack-tox-functional http://logs.openstack.org/46/670746/2/check/openstack-tox-functional/92277fc/ : SUCCESS in 8m 54s\n- openstack-tox-functional-py35 http://logs.openstack.org/46/670746/2/check/openstack-tox-functional-py35/9cc069b/ : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a3c89b9cf617e155da7f90c8e1b75cc32d14ca5a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-15 02:12:43.000000000","message":"Abandoned\n\nAlready being addressed:\n\nhttps://review.opendev.org/#/c/664475/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8b121d340275af2f9180de78599e01cb0bd5e4f","revisions":{"20f2c7af00027344665ea02cc4ce4d05b06a74c4":{"kind":"REWORK","_number":1,"created":"2019-07-15 01:12:22.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/46/670746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/670746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/670746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/670746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/670746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/670746/1"}}},"commit":{"parents":[{"commit":"d319acbd649720bc459e929eaa7ee04c3acf0a86","subject":"Update the glance installation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-15 01:10:29.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-15 01:12:18.000000000","tz":480},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in\nplace\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I28b01dc028427bc08609d2f30031bda547a4c075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20f2c7af00027344665ea02cc4ce4d05b06a74c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20f2c7af00027344665ea02cc4ce4d05b06a74c4"}]},"branch":"refs/heads/master"},"c8b121d340275af2f9180de78599e01cb0bd5e4f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-15 01:15:42.000000000","uploader":{"_account_id":30395,"name":"Arthur Dayne","display_name":"Arthur Dayne","email":"palagend@163.com","username":"acewit"},"ref":"refs/changes/46/670746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/670746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/670746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/670746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/670746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/670746/2"}}},"commit":{"parents":[{"commit":"d319acbd649720bc459e929eaa7ee04c3acf0a86","subject":"Update the glance installation page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d319acbd649720bc459e929eaa7ee04c3acf0a86"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-15 01:10:29.000000000","tz":480},"committer":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-07-15 01:14:40.000000000","tz":480},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I28b01dc028427bc08609d2f30031bda547a4c075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8b121d340275af2f9180de78599e01cb0bd5e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8b121d340275af2f9180de78599e01cb0bd5e4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
