)]}'
{"id":"openstack%2Fglance~671707","triplet_id":"openstack%2Fglance~master~Ifbd6a93cce21ea18d85039b9a8b1508de61deeda","project":"openstack/glance","branch":"master","topic":"1837200","hashtags":[],"change_id":"Ifbd6a93cce21ea18d85039b9a8b1508de61deeda","subject":"glance-manage support to obfuscate deleted images","status":"ABANDONED","created":"2019-07-19 11:23:19.000000000","updated":"2021-07-07 14:42:56.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2737ca6dc08f1648bc331d134956a438fd336e70","_number":671707,"virtual_id_number":671707,"owner":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-08-28 23:58:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":1,"date":"2019-09-10 14:11:14.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 14:26:53.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-28 23:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-10 14:11:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"60d21eb85be20f50fec6b470fda13fb9138a7a9c","author":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"date":"2019-07-19 11:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49c3ae4a9a6c5c3ae55918cfbd4533e0c067a904","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-19 12:15:47.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/07/671707/1/check-not-vote/ext-glance-zuul/dea7e20 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cc6fe30a622c19f800628ca7ffdc3903eb50c38f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:57:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/07/671707/1/check/tempest-full/108c91d/ : SUCCESS in 1h 32m 58s\n- neutron-grenade http://logs.openstack.org/07/671707/1/check/neutron-grenade/309b357/ : SUCCESS in 58m 40s\n- grenade-py3 http://logs.openstack.org/07/671707/1/check/grenade-py3/8eadea2/ : FAILURE in 47m 37s\n- tempest-full-py3 http://logs.openstack.org/07/671707/1/check/tempest-full-py3/0a2ec7d/ : SUCCESS in 1h 22m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671707/1/check/openstack-tox-lower-constraints/1d0574b/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/07/671707/1/check/openstack-tox-pep8/e62a823/ : FAILURE in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/07/671707/1/check/openstack-tox-py27/17d3540/ : SUCCESS in 6m 34s\n- openstack-tox-py35 http://logs.openstack.org/07/671707/1/check/openstack-tox-py35/d698ef0/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/07/671707/1/check/openstack-tox-py36/e79d77b/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/07/671707/1/check/openstack-tox-py37/27b1491/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/07/671707/1/check/openstack-tox-docs/65ff141/html/ : SUCCESS in 4m 53s\n- openstack-tox-functional http://logs.openstack.org/07/671707/1/check/openstack-tox-functional/8330939/ : FAILURE in 10m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/07/671707/1/check/openstack-tox-functional-py35/14c02d9/ : SUCCESS in 11m 35s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/07/671707/1/check/glance-eventlet-ssl-handshake-broken-py3/4c9a3b3/ : FAILURE in 3m 55s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/07/671707/1/check/glance-code-constants-check/4f22dd7/ : SUCCESS in 5m 29s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/07/671707/1/check/devstack-plugin-ceph-tempest/f86a645/ : FAILURE in 1h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"329e7b134385546ad812fe725c0e11cff54de3fe","author":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"date":"2019-07-19 13:58:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7c0f29204321cc9e97a5ee5c079cdb3038d01de","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-19 14:37:54.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/07/671707/2/check-not-vote/ext-glance-zuul/994fd50 : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4d0d78a551c734817b461c9ce76a0e2801e7d968","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 19:06:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/07/671707/2/check/tempest-full/97f9dff/ : SUCCESS in 1h 37m 18s\n- neutron-grenade http://logs.openstack.org/07/671707/2/check/neutron-grenade/296e0a1/ : SUCCESS in 57m 57s\n- grenade-py3 http://logs.openstack.org/07/671707/2/check/grenade-py3/7b32f67/ : SUCCESS in 59m 38s\n- tempest-full-py3 http://logs.openstack.org/07/671707/2/check/tempest-full-py3/05e378b/ : SUCCESS in 1h 23m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671707/2/check/openstack-tox-lower-constraints/26f68e1/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/07/671707/2/check/openstack-tox-pep8/0d1f81f/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/07/671707/2/check/openstack-tox-py27/c589c68/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/07/671707/2/check/openstack-tox-py35/565a4b2/ : SUCCESS in 7m 48s\n- openstack-tox-py36 http://logs.openstack.org/07/671707/2/check/openstack-tox-py36/dab576d/ : SUCCESS in 6m 30s\n- openstack-tox-py37 http://logs.openstack.org/07/671707/2/check/openstack-tox-py37/ad14688/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/07/671707/2/check/openstack-tox-docs/1d70c25/html/ : SUCCESS in 5m 24s\n- openstack-tox-functional http://logs.openstack.org/07/671707/2/check/openstack-tox-functional/3948415/ : SUCCESS in 11m 05s\n- openstack-tox-functional-py35 http://logs.openstack.org/07/671707/2/check/openstack-tox-functional-py35/f4c07bf/ : SUCCESS in 12m 16s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/07/671707/2/check/glance-eventlet-ssl-handshake-broken-py3/332845b/ : FAILURE in 5m 14s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/07/671707/2/check/glance-code-constants-check/9f5685c/ : SUCCESS in 5m 00s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/07/671707/2/check/devstack-plugin-ceph-tempest/92714c9/ : SUCCESS in 1h 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf1122dd697071656297864cc6a715c78e5a35b7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-20 02:03:19.000000000","message":"Patch Set 2:\n\nLooks good at first glance.  Haven\u0027t had time to test it yet.  Will need a release note -- see https://docs.openstack.org/glance/latest/contributor/release-notes.html if you haven\u0027t done one before.","accounts_in_message":[],"_revision_number":2},{"id":"c226d882b9694cd03d1ef62eb2ed780176cdee69","author":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"date":"2019-07-23 13:45:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cacfa8520643d0f215616c382153007eb1fe6899","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-23 14:26:53.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/07/671707/3/check-not-vote/ext-glance-zuul/c078f8b : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":3},{"id":"8246998180257bfc4f5a177227dd1981d43e65ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:19:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/07/671707/3/check/tempest-full/ee19b9c/ : SUCCESS in 1h 32m 04s\n- neutron-grenade http://logs.openstack.org/07/671707/3/check/neutron-grenade/355c644/ : SUCCESS in 1h 11m 46s\n- grenade-py3 http://logs.openstack.org/07/671707/3/check/grenade-py3/49d3a79/ : SUCCESS in 1h 08m 59s\n- tempest-full-py3 http://logs.openstack.org/07/671707/3/check/tempest-full-py3/f3fa50c/ : SUCCESS in 1h 33m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671707/3/check/openstack-tox-lower-constraints/c741ac2/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/07/671707/3/check/openstack-tox-pep8/b972ae3/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/07/671707/3/check/openstack-tox-py27/e89067f/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/07/671707/3/check/openstack-tox-py35/133fe20/ : SUCCESS in 7m 45s\n- openstack-tox-py36 http://logs.openstack.org/07/671707/3/check/openstack-tox-py36/e684780/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/07/671707/3/check/openstack-tox-py37/e1bd612/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/07/671707/3/check/openstack-tox-docs/2be01b9/html/ : SUCCESS in 7m 42s\n- build-openstack-releasenotes http://logs.openstack.org/07/671707/3/check/build-openstack-releasenotes/207f78a/html/ : SUCCESS in 7m 09s\n- openstack-tox-functional http://logs.openstack.org/07/671707/3/check/openstack-tox-functional/99b5bb7/ : SUCCESS in 9m 21s\n- openstack-tox-functional-py35 http://logs.openstack.org/07/671707/3/check/openstack-tox-functional-py35/ed75de5/ : SUCCESS in 9m 40s\n- glance-eventlet-ssl-handshake-broken-py3 http://logs.openstack.org/07/671707/3/check/glance-eventlet-ssl-handshake-broken-py3/3d7ce78/ : FAILURE in 5m 22s (non-voting)\n- glance-code-constants-check http://logs.openstack.org/07/671707/3/check/glance-code-constants-check/c80d99a/ : SUCCESS in 4m 38s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/07/671707/3/check/devstack-plugin-ceph-tempest/320856e/ : SUCCESS in 1h 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b838ddb7aefc97eef0a070fcd2e58432c9bae79c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-25 01:17:08.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"d68398695879b36fcdb96a39c3431f491bf40533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 03:58:50.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- glance-tox-py27-glance_store-tips http://logs.openstack.org/07/671707/3/experimental/glance-tox-py27-glance_store-tips/071af37/ : SUCCESS in 5m 32s\n- glance-tox-py36-glance_store-tips http://logs.openstack.org/07/671707/3/experimental/glance-tox-py36-glance_store-tips/1e61a22/ : SUCCESS in 4m 25s\n- glance-tox-functional-glance_store-tips http://logs.openstack.org/07/671707/3/experimental/glance-tox-functional-glance_store-tips/2651b8a/ : FAILURE in 9m 24s\n- glance-tox-functional-py36-glance_store-tips http://logs.openstack.org/07/671707/3/experimental/glance-tox-functional-py36-glance_store-tips/321b2ed/ : FAILURE in 8m 40s\n- barbican-simple-crypto-devstack-tempest http://logs.openstack.org/07/671707/3/experimental/barbican-simple-crypto-devstack-tempest/00bf9d0/ : SUCCESS in 37m 34s\n- glance-dsvm-grenade-multinode http://logs.openstack.org/07/671707/3/experimental/glance-dsvm-grenade-multinode/715ad36/ : SUCCESS in 1h 04m 04s\n- devstack-plugin-ceph-tempest-py3 http://logs.openstack.org/07/671707/3/experimental/devstack-plugin-ceph-tempest-py3/3f1ac66/ : SUCCESS in 1h 03m 40s\n- tempest-pg-full http://logs.openstack.org/07/671707/3/experimental/tempest-pg-full/5c53187/ : SUCCESS in 1h 28m 21s\n- tempest-full-py3-opensuse150 http://logs.openstack.org/07/671707/3/experimental/tempest-full-py3-opensuse150/ebef6da/ : TIMED_OUT in 2h 05m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6eff847f533b0a354dc5d9330f2c433d17786f69","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-07-25 11:33:32.000000000","message":"Patch Set 3:\n\n@Belmiro: ignore the experimental failures, I was using your patch to check the state of glance_store (completely unrelated to this patch).","accounts_in_message":[],"_revision_number":3},{"id":"6222f23185580931ccec9c5b7ca5bdbb63aaf0db","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-15 16:11:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)\n\nI think the basic code is ok, though I noticed a few things to address.","accounts_in_message":[],"_revision_number":3},{"id":"0eb35819b737b2bd83880e739a0d64b1470e8dd0","author":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"date":"2019-08-19 14:22:05.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nThanks for the review. I will update it asap.","accounts_in_message":[],"_revision_number":3},{"id":"c56a86b8884e71a8fcb76e424cf83be6de835287","author":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"date":"2019-08-28 22:07:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6323615d9b8d93ceeaf2f4c40f1bfe25f0e17f26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 23:58:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b8e87c35af6246f5896aafb03a934ef4 : SUCCESS in 1h 44m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/03300397a3244faaae43733ddc756438 : SUCCESS in 1h 06m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09ea2efa7ab24e6a8624f5a9511baabe : SUCCESS in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ce9cc6b93e44c3296475010b184482f : SUCCESS in 1h 30m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2200cd16aadf4d0a95352a3a3d34b29c : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e766e1f393584a2cb93f434481ac3b95 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ef160c163ef4b53918bf8bdbeadb6fe : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/667df2fe87eb4655a14e135a8c6f6a0e : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/527d44c576b14cea8b4f6d1851ad441b : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7bb23f702d14cb49aa8e9d6faf1a1c0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1bd268ba13456991ec993837ef56e5 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faf65e76e0c24ab3bd3f5c7a7f6434d7 : SUCCESS in 10m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb9591cd82734f0597ac78e8caf1c4a2 : SUCCESS in 9m 01s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a7a758bb28104f89939fbe646db4bf20 : SUCCESS in 9m 18s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/cdc85a2d5ddc44b4a28255244d9893ac : FAILURE in 4m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c2813c3635894e7b9e28c9a44f0a1e0c : SUCCESS in 4m 39s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/474f441d27564befaabef58eda5eed4b : SUCCESS in 1h 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"417b4da5cbb38200842a787b611bd7555e4ab266","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-10 14:11:14.000000000","message":"Patch Set 4: Review-Priority+1\n\nEven though this is a bug, it would be good to beat the feature freeze.","accounts_in_message":[],"_revision_number":4},{"id":"2737ca6dc08f1648bc331d134956a438fd336e70","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-07 14:42:56.000000000","message":"Abandoned\n\nI believe this was fixed with\nhttps://review.opendev.org/c/openstack/glance/+/579507\n\n\nSee also https://specs.openstack.org/openstack/glance-specs/specs/rocky/implemented/glance/mitigate-ossn-0075.html","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"440d60400d8becb31d3f09032343e5c030a734fb","revisions":{"e04227fd429263b23ceb4951aef13a8f1f7a32a8":{"kind":"REWORK","_number":1,"created":"2019-07-19 11:23:19.000000000","uploader":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"ref":"refs/changes/07/671707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/671707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/671707/1"}}},"commit":{"parents":[{"commit":"a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0"}]}],"author":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-19 11:14:21.000000000","tz":120},"committer":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-19 11:14:21.000000000","tz":120},"subject":"glance-manage support to obfuscate deleted images","message":"glance-manage support to obfuscate deleted images\n\nIntroduces a glance-manage command to obfuscate deleted images from\nimages table that can\u0027t be purged because OSSN-0075.\n\nThe image id is preserved however all the other fields related to\nthe image are obfuscated. All the dates are set to glance initial\ncommit (values required) and all the other fields are set to NULL\nexcept protected, visibility and os_hidden.\n\nProtected and os_hidden are set to true and visibility to private.\n\nChange-Id: Ifbd6a93cce21ea18d85039b9a8b1508de61deeda\nCloses-Bug: #1837200\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e04227fd429263b23ceb4951aef13a8f1f7a32a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e04227fd429263b23ceb4951aef13a8f1f7a32a8"}]},"branch":"refs/heads/master"},"0a3f7b676c4f47e4bce63ce8ddb35674de47bed7":{"kind":"REWORK","_number":2,"created":"2019-07-19 13:58:47.000000000","uploader":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"ref":"refs/changes/07/671707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/671707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/671707/2"}}},"commit":{"parents":[{"commit":"a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0"}]}],"author":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-19 11:14:21.000000000","tz":120},"committer":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-19 13:56:51.000000000","tz":120},"subject":"glance-manage support to obfuscate deleted images","message":"glance-manage support to obfuscate deleted images\n\nIntroduces a glance-manage command to obfuscate deleted images from\nimages table that can\u0027t be purged because OSSN-0075.\n\nThe image id is preserved however all the other fields related to\nthe image are obfuscated. All the dates are set to glance initial\ncommit (values required) and all the other fields are set to NULL\nexcept protected, visibility and os_hidden.\n\nProtected and os_hidden are set to true and visibility to private.\n\nChange-Id: Ifbd6a93cce21ea18d85039b9a8b1508de61deeda\nCloses-Bug: #1837200\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a3f7b676c4f47e4bce63ce8ddb35674de47bed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a3f7b676c4f47e4bce63ce8ddb35674de47bed7"}]},"branch":"refs/heads/master"},"eb79c356aac364454381f1c2e8f9b0ad9cec7733":{"kind":"REWORK","_number":3,"created":"2019-07-23 13:45:29.000000000","uploader":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"ref":"refs/changes/07/671707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/671707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/671707/3"}}},"commit":{"parents":[{"commit":"a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0"}]}],"author":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-19 11:14:21.000000000","tz":120},"committer":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-23 13:42:16.000000000","tz":0},"subject":"glance-manage support to obfuscate deleted images","message":"glance-manage support to obfuscate deleted images\n\nIntroduces a glance-manage command to obfuscate deleted images from\nimages table that can\u0027t be purged because OSSN-0075.\n\nThe image id is preserved however all the other fields related to\nthe image are obfuscated. All the dates are set to glance initial\ncommit (values required) and all the other fields are set to NULL\nexcept protected, visibility and os_hidden.\n\nProtected and os_hidden are set to true and visibility to private.\n\nChange-Id: Ifbd6a93cce21ea18d85039b9a8b1508de61deeda\nCloses-Bug: #1837200\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb79c356aac364454381f1c2e8f9b0ad9cec7733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb79c356aac364454381f1c2e8f9b0ad9cec7733"}]},"branch":"refs/heads/master"},"440d60400d8becb31d3f09032343e5c030a734fb":{"kind":"REWORK","_number":4,"created":"2019-08-28 22:07:12.000000000","uploader":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"ref":"refs/changes/07/671707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/671707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/671707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/671707/4"}}},"commit":{"parents":[{"commit":"a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0","subject":"Merge \"Bump openstackdocstheme to 1.20.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7b6d1eaa81d801eacf45332955fe4f2b52d0bc0"}]}],"author":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-07-19 11:14:21.000000000","tz":120},"committer":{"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","date":"2019-08-28 22:07:10.000000000","tz":120},"subject":"glance-manage support to obfuscate deleted images","message":"glance-manage support to obfuscate deleted images\n\nIntroduces a glance-manage command to obfuscate deleted images from\nimages table that can\u0027t be purged because OSSN-0075.\n\nThe image id is preserved however all the other fields related to\nthe image are obfuscated. All the dates are set to glance initial\ncommit (values required) and all the other fields are set to NULL\nexcept protected, visibility and os_hidden.\n\nProtected and os_hidden are set to true and visibility to private.\n\nChange-Id: Ifbd6a93cce21ea18d85039b9a8b1508de61deeda\nCloses-Bug: #1837200\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/440d60400d8becb31d3f09032343e5c030a734fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/440d60400d8becb31d3f09032343e5c030a734fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
