)]}'
{"id":"openstack%2Fglance~672098","triplet_id":"openstack%2Fglance~master~I17c97cb7655bdedc3927f4883b45ba4dcfc55516","project":"openstack/glance","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"I17c97cb7655bdedc3927f4883b45ba4dcfc55516","subject":"Update api-ref location","status":"MERGED","created":"2019-07-22 16:46:25.000000000","updated":"2019-07-29 07:53:46.000000000","submitted":"2019-07-29 07:53:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672098-1564386826288-d14e3b94","meta_rev_id":"44d09df58a07c3944d9def08de1d038242fbb9d0","_number":672098,"virtual_id_number":672098,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-22 20:32:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-22 18:36:39.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-07-29 07:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-29 05:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2019-07-23 07:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-07-29 05:04:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 16:59:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-22 20:32:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-23 07:54:13.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-07-29 05:04:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-07-29 07:53:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cfa5223080e2c1ca61d56e79d36b6a22be3806c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 16:46:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e6d6735eed6984ab3754c32a51222c9397d480a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-22 16:59:33.000000000","message":"Patch Set 1:\n\nThe problem with updating the release notes docs is that they will now show up in the latest generated notes.","accounts_in_message":[],"_revision_number":1},{"id":"b18aa3cbf937e12b1867d62a25750a8474219d12","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 17:08:04.000000000","message":"Patch Set 1: Workflow-1\n\nLet me check the releasenotes output ...","accounts_in_message":[],"_revision_number":1},{"id":"2a704b1aae053d8d462854823feb1b7803654e19","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 17:15:21.000000000","message":"Patch Set 1:\n\nremoved releasenotes","accounts_in_message":[],"_revision_number":1},{"id":"d489d588eb6cff1272f68a158a5f8a2a371479c5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 17:15:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"526fcc0017ab9ba11d2ade0c3eaebbb5a05a8a72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:27:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/98/672098/2/check/test-release-openstack/5f9bbce/ : SUCCESS in 2m 52s\n- build-openstack-api-ref http://logs.openstack.org/98/672098/2/check/build-openstack-api-ref/49432fc/html/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/672098/2/check/openstack-tox-lower-constraints/fc5c906/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/98/672098/2/check/openstack-tox-pep8/5e25226/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/98/672098/2/check/openstack-tox-py27/580721d/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/98/672098/2/check/openstack-tox-py35/af973c4/ : FAILURE in 6m 14s\n- openstack-tox-py36 http://logs.openstack.org/98/672098/2/check/openstack-tox-py36/8616596/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/98/672098/2/check/openstack-tox-py37/06017aa/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/98/672098/2/check/openstack-tox-docs/74163fd/html/ : SUCCESS in 5m 44s\n- build-openstack-releasenotes http://logs.openstack.org/98/672098/2/check/build-openstack-releasenotes/ac684ee/html/ : SUCCESS in 9m 44s\n- openstack-tox-functional http://logs.openstack.org/98/672098/2/check/openstack-tox-functional/13b676c/ : SUCCESS in 8m 15s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/672098/2/check/openstack-tox-functional-py35/fe7e907/ : TIMED_OUT in 31m 21s","accounts_in_message":[],"_revision_number":2},{"id":"153e59a8a38b32ab14ae2ef58d25637d5b331d8a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 18:36:39.000000000","message":"Patch Set 2:\n\nrecheck - timeout","accounts_in_message":[],"_revision_number":2},{"id":"371eda0e72d90cbf39956838d807780e26e9a3d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:57:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/672098/2/check/test-release-openstack/9a25dc7/ : SUCCESS in 3m 22s\n- build-openstack-api-ref http://logs.openstack.org/98/672098/2/check/build-openstack-api-ref/8fb9bb4/html/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/672098/2/check/openstack-tox-lower-constraints/3ed4628/ : SUCCESS in 6m 09s\n- openstack-tox-pep8 http://logs.openstack.org/98/672098/2/check/openstack-tox-pep8/2946079/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/98/672098/2/check/openstack-tox-py27/3ab2983/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/98/672098/2/check/openstack-tox-py35/9a735a8/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/98/672098/2/check/openstack-tox-py36/e4a7a4c/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/98/672098/2/check/openstack-tox-py37/df4170f/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/98/672098/2/check/openstack-tox-docs/0f120e2/html/ : SUCCESS in 5m 46s\n- build-openstack-releasenotes http://logs.openstack.org/98/672098/2/check/build-openstack-releasenotes/28a65dc/html/ : SUCCESS in 9m 12s\n- openstack-tox-functional http://logs.openstack.org/98/672098/2/check/openstack-tox-functional/d11d80b/ : SUCCESS in 8m 32s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/672098/2/check/openstack-tox-functional-py35/29c01f0/ : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":2},{"id":"cb996eb3a97aa3c4772c8a97a1559cb0cba80321","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-22 20:32:13.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/98/672098/2/check-not-vote/ext-glance-zuul/e01ec9d : SUCCESS in 40m 33s","accounts_in_message":[],"_revision_number":2},{"id":"deffa83efcd0a7d88a22026266c6abc012dd7f4a","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-07-23 07:54:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"799544395d87a847003b457eccef3b1513e8a4b3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-29 05:04:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nApproving as per glance doc policy.\n\nTank you Andreas!!!","accounts_in_message":[],"_revision_number":2},{"id":"4b8b48bc063d27127faa701f6a22803921ddeca8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:05:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e046c86eccc0b37894b8297ebda7a61da97c428","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:18:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/98/672098/2/gate/test-release-openstack/8ac4c95/ : SUCCESS in 4m 32s\n- build-openstack-api-ref http://logs.openstack.org/98/672098/2/gate/build-openstack-api-ref/902edbf/html/ : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/672098/2/gate/openstack-tox-lower-constraints/32dea4d/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/98/672098/2/gate/openstack-tox-pep8/0159dae/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py27/9983943/ : SUCCESS in 5m 09s\n- openstack-tox-py35 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py35/3fa6fff/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py36/1d09950/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py37/856cbc4/ : FAILURE in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/98/672098/2/gate/openstack-tox-docs/033ca71/html/ : SUCCESS in 4m 21s\n- build-openstack-releasenotes http://logs.openstack.org/98/672098/2/gate/build-openstack-releasenotes/8fc4681/html/ : SUCCESS in 7m 44s\n- openstack-tox-functional http://logs.openstack.org/98/672098/2/gate/openstack-tox-functional/25be3ab/ : SUCCESS in 8m 11s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/672098/2/gate/openstack-tox-functional-py35/bcb3016/ : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c8e84a32130ecfcbd6dfc9929543c514fbf65cff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-07-29 05:26:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e841a36016cbee3e271cec3ef3a50a434dc1c154","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:40:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/672098/2/check/test-release-openstack/402dfdd/ : SUCCESS in 4m 13s\n- build-openstack-api-ref http://logs.openstack.org/98/672098/2/check/build-openstack-api-ref/216638e/html/ : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/672098/2/check/openstack-tox-lower-constraints/cf6ac87/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/98/672098/2/check/openstack-tox-pep8/8231d0b/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/98/672098/2/check/openstack-tox-py27/86289b5/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/98/672098/2/check/openstack-tox-py35/babafdb/ : SUCCESS in 5m 50s\n- openstack-tox-py36 http://logs.openstack.org/98/672098/2/check/openstack-tox-py36/8b2417d/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/98/672098/2/check/openstack-tox-py37/9131f76/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/98/672098/2/check/openstack-tox-docs/d100bc9/html/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/98/672098/2/check/build-openstack-releasenotes/e7f89d1/html/ : SUCCESS in 8m 04s\n- openstack-tox-functional http://logs.openstack.org/98/672098/2/check/openstack-tox-functional/5078e95/ : SUCCESS in 10m 55s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/672098/2/check/openstack-tox-functional-py35/bf4c004/ : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c0064a9cad40bc64c6ac3aa6dcdb631c2e650844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:41:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7951d66b029e1f8a268cd951fe5f8f2375fd6a27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 07:53:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/672098/2/gate/test-release-openstack/378182e/ : SUCCESS in 3m 18s\n- build-openstack-api-ref http://logs.openstack.org/98/672098/2/gate/build-openstack-api-ref/60e2e05/html/ : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/98/672098/2/gate/openstack-tox-lower-constraints/b2ca182/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/98/672098/2/gate/openstack-tox-pep8/08d3b7b/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py27/d4dccde/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py35/3c08099/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py36/2e302c0/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/98/672098/2/gate/openstack-tox-py37/0e668a0/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/98/672098/2/gate/openstack-tox-docs/ad29e9d/html/ : SUCCESS in 4m 26s\n- build-openstack-releasenotes http://logs.openstack.org/98/672098/2/gate/build-openstack-releasenotes/f78367d/html/ : SUCCESS in 8m 13s\n- openstack-tox-functional http://logs.openstack.org/98/672098/2/gate/openstack-tox-functional/3473350/ : SUCCESS in 8m 40s\n- openstack-tox-functional-py35 http://logs.openstack.org/98/672098/2/gate/openstack-tox-functional-py35/84a3751/ : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a7742e3fd6f502d03bc321793709ab5e612d98b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 07:53:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15609a11bd66b3586b2ac3271fd4c1d36a414409","revisions":{"65014fe9cc9bb11b240d562b622e6e741f59caed":{"kind":"REWORK","_number":1,"created":"2019-07-22 16:46:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/672098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/672098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/672098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/672098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/672098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/672098/1"}}},"commit":{"parents":[{"commit":"9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a","subject":"Merge \"MultiStore: Avoid potential KeyError in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:46:16.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:46:16.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: I17c97cb7655bdedc3927f4883b45ba4dcfc55516\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65014fe9cc9bb11b240d562b622e6e741f59caed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65014fe9cc9bb11b240d562b622e6e741f59caed"}]},"branch":"refs/heads/master"},"15609a11bd66b3586b2ac3271fd4c1d36a414409":{"kind":"REWORK","_number":2,"created":"2019-07-22 17:15:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/98/672098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/672098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/672098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/672098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/672098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/672098/2"}}},"commit":{"parents":[{"commit":"9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a","subject":"Merge \"MultiStore: Avoid potential KeyError in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9bb432c30d8a4be7ce82f5c6d79f85f9c4355a3a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:46:16.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 17:15:05.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: I17c97cb7655bdedc3927f4883b45ba4dcfc55516\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15609a11bd66b3586b2ac3271fd4c1d36a414409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15609a11bd66b3586b2ac3271fd4c1d36a414409"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
