)]}'
{"id":"openstack%2Fglance~675182","triplet_id":"openstack%2Fglance~master~I2f4e7ef4f816afbd0524a6265d5605afa1da5f86","project":"openstack/glance","branch":"master","topic":"bp/libvirt-pmu-configuration","hashtags":[],"change_id":"I2f4e7ef4f816afbd0524a6265d5605afa1da5f86","subject":"add hw_pmu metadef","status":"MERGED","created":"2019-08-07 19:09:10.000000000","updated":"2019-09-06 03:15:14.000000000","submitted":"2019-09-06 03:13:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675182-1567739628101-72f0986f","meta_rev_id":"1d706e73bf0c1f9a95d71231892e68e834f1a524","_number":675182,"virtual_id_number":675182,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":2,"date":"2019-09-06 03:13:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-06 02:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-09-04 11:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-08-22 13:24:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-08-28 03:01:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 02:37:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 20:41:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 13:24:07.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-08-28 03:01:35.000000000","updated_by":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"reviewer":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"state":"REVIEWER"},{"updated":"2019-09-04 11:05:40.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-09-06 02:37:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-06 03:13:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fb0a257eaf48d31a682c21763db6920fdfa84a2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-07 19:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4be7c125859114a19eca7b86f636fd720e278fcb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-07 19:18:07.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/82/675182/1/check-not-vote/ext-glance-zuul/8cb353a : FAILURE in 1m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ad7cd0a450f871707406609ae4edee009bf96ca2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-07 19:31:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nI think there\u0027s a typo in the metadef description.  Thanks for adding the release note.","accounts_in_message":[],"_revision_number":1},{"id":"ce0ee35bdc6abdeaa59701fcc3b3144d927ec67e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 20:04:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/82/675182/1/check/openstack-tox-lower-constraints/6466eeb/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://logs.opendev.org/82/675182/1/check/openstack-tox-pep8/cdaed4c/ : SUCCESS in 5m 17s\n- openstack-tox-py27 https://logs.opendev.org/82/675182/1/check/openstack-tox-py27/f3467f5/ : SUCCESS in 6m 01s\n- openstack-tox-py35 https://logs.opendev.org/82/675182/1/check/openstack-tox-py35/4cdb690/ : SUCCESS in 5m 38s\n- openstack-tox-py36 https://logs.opendev.org/82/675182/1/check/openstack-tox-py36/9827691/ : SUCCESS in 6m 15s\n- openstack-tox-py37 https://logs.opendev.org/82/675182/1/check/openstack-tox-py37/fe6de21/ : SUCCESS in 6m 37s\n- openstack-tox-docs https://logs.opendev.org/82/675182/1/check/openstack-tox-docs/b58e7b4/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://logs.opendev.org/82/675182/1/check/build-openstack-releasenotes/76c398f/html/ : SUCCESS in 9m 04s\n- openstack-tox-functional https://logs.opendev.org/82/675182/1/check/openstack-tox-functional/e073f71/ : SUCCESS in 8m 50s\n- openstack-tox-functional-py35 https://logs.opendev.org/82/675182/1/check/openstack-tox-functional-py35/2d418e4/ : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":1},{"id":"07869332f053d9a34ce35cfe3b451f7e7d16d94e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-07 20:19:00.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f303cb52812c58d652ec349d002534ee5aed24c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-07 20:21:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"866e7ca239164779fb34f5d47ed59e7c9f078465","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-07 20:25:21.000000000","message":"Patch Set 2:\n\ngeneral question by the way.\ni notice that there are a few other image properties that\nnova support but are not reflected in the metadefs.\n\nthe two patch i have up are for sperate feature that are being added in train so i have made them as spereate patchs but if i want to add metadefs for some of undocumeted image properties we support i assume it would be fine to group them in a larger patch rather then haveing a patch per feature.","accounts_in_message":[],"_revision_number":2},{"id":"9220cdf8eb5a23441be1fb1c00745f3b33de21f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 20:36:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/82/675182/2/check/openstack-tox-lower-constraints/24eee03/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://logs.opendev.org/82/675182/2/check/openstack-tox-pep8/c54078f/ : SUCCESS in 5m 40s\n- openstack-tox-py27 https://logs.opendev.org/82/675182/2/check/openstack-tox-py27/332fd80/ : SUCCESS in 4m 56s\n- openstack-tox-py35 https://logs.opendev.org/82/675182/2/check/openstack-tox-py35/c9ccd66/ : SUCCESS in 4m 59s\n- openstack-tox-py36 https://logs.opendev.org/82/675182/2/check/openstack-tox-py36/e81c0a1/ : SUCCESS in 5m 31s\n- openstack-tox-py37 https://logs.opendev.org/82/675182/2/check/openstack-tox-py37/80e12d2/ : SUCCESS in 5m 28s\n- openstack-tox-docs https://logs.opendev.org/82/675182/2/check/openstack-tox-docs/f035dea/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://logs.opendev.org/82/675182/2/check/build-openstack-releasenotes/5567941/html/ : SUCCESS in 8m 52s\n- openstack-tox-functional https://logs.opendev.org/82/675182/2/check/openstack-tox-functional/442fe75/ : SUCCESS in 8m 16s\n- openstack-tox-functional-py35 https://logs.opendev.org/82/675182/2/check/openstack-tox-functional-py35/94d8ef3/ : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4f146b03ffae5fe63a0361f281def4f1c5b2b85d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-07 20:41:58.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/82/675182/2/check-not-vote/ext-glance-zuul/c877542 : FAILURE in 1m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6b56c09e5bdc3e6542c73160b4b5185f9fcece38","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-08 02:35:26.000000000","message":"Patch Set 2:\n\n@sean: yes, I think a general patch to bring the compute metadefs up to date would be fine.  It would be great if you could put one up.  Also, for the properties that go on images, we keep a doc page to supplement the metadefs, but it\u0027s probably missing some properties as well.  It\u0027s doc/source/admin/useful-image-properties.rst","accounts_in_message":[],"_revision_number":2},{"id":"f4f43fce1300da169b1d777ebb43f3bcd4e74703","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-08 02:39:05.000000000","message":"Patch Set 2: Code-Review-1\n\nWhich makes me realize that hw_pmu should be added to that doc.  Sorry about that.  It would probably be best to add it as part of this patch to keep the changes together.  doc/source/admin/useful-image-properties.rst","accounts_in_message":[],"_revision_number":2},{"id":"465868d6baf5bed77202471d00e9090472bf4d79","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-12 14:52:51.000000000","message":"Patch Set 2:\n\ncool ill update that too. -w until that is done","accounts_in_message":[],"_revision_number":2},{"id":"09b464cb45954044f965d8a9925751f93619ec9d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-15 19:00:03.000000000","message":"Patch Set 2: Code-Review-1\n\nYou should update the docs as well:\n\nhttps://docs.openstack.org/glance/latest/admin/useful-image-properties.html","accounts_in_message":[],"_revision_number":2},{"id":"08045c15c774d7d66c920bbbe9c2d18b8aa8be1f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-20 14:44:28.000000000","message":"Patch Set 2:\n\nHeh, now I see that Brian said the same thing. I\u0027m just piling on. :)","accounts_in_message":[],"_revision_number":2},{"id":"b26639aae6feddd12ae54d36a95627b46a06ce7a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-20 23:11:50.000000000","message":"Patch Set 2: Workflow-1\n\nforgot to -w\nthe nova patch is currently in the gate so im going to update this now","accounts_in_message":[],"_revision_number":2},{"id":"53753a0a5069275545c28ecf1cc2fd2102d51fa9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-21 00:06:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7a96b96a2cf488e2ea853ebe49f69739330c3449","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 00:22:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed730ac925f1479fbfbad3dea70d6951 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db46f7f61c554e748fee817c1fd0c34c : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f7c265be51247558b042205970ccbb4 : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9fc18a06686540fdb672b41f3b5f7813 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50bef382a24c438c868cdcb3bac127f9 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f431f707ec564d349aed59f6c44a399f : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b719b8df0f412d943e1314e068704b : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeeb104eb38248bb86d71634a33b45bf : SUCCESS in 9m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b2db3de2db194036872f9102dd496667 : SUCCESS in 9m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/68692491ace249518f3b1485c7104047 : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":3},{"id":"ee15a0e74658d44d04a9122d3ca849cba3c02454","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-22 13:24:07.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"03f9d5d9951d5c16e6fef1730b2843ca14d0c03d","author":{"_account_id":26233,"name":"zengjia","email":"zengjia@awcloud.com","username":"zengjia87"},"date":"2019-08-28 03:01:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9172418aa7e98e8af2b0b2c826d1146470773df7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-04 11:05:40.000000000","message":"Patch Set 3: Code-Review+2\n\nThis looks good, Thanks for the docs and reno, definitely makes my life easier seeing it all there together.","accounts_in_message":[],"_revision_number":3},{"id":"21654737ec38d926d1715635d3a249721715b8d5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-06 02:37:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks good.  Thanks, Sean.","accounts_in_message":[],"_revision_number":3},{"id":"bfcacbb2124c2ee6ec41c42cd999e77fc7d1c055","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 02:37:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"98357b10549276c149ff15e140110e25f4c78d66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:13:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5be05bf1a76940508b0bc7995cb6c1e0 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dee163c1e0345a0bcb450e93ae7095a : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3f4cc83ff1f4857a117b8dfb103ced6 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5458731e60894d038947aea385e0944d : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff7466c3f25a476c82fe964f7870cc9a : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040c867e206e4f4288604fcc7346f431 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80a293d6b284fb4aa4c73e4b4e8a672 : SUCCESS in 10m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f27cfb9ddc974b658584fd2441db4939 : SUCCESS in 9m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bbc5c670f8c046289d8c91106cdc070d : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7b8d71bcc50734d611714c90a9e98f0a6d01a20e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:13:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"15b4b1b6862a3e642faada4b862c60f8eadf13ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:15:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3141bb7f1005417a824b73fc8afe8850 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2998c795ed5f1accf041373c0b4cbf0883357862","revisions":{"fcb92b9978d80983ca3df4f0eada176ec554de82":{"kind":"REWORK","_number":1,"created":"2019-08-07 19:09:10.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/82/675182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/675182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/675182/1"}}},"commit":{"parents":[{"commit":"4915fb8793a89374fb5a881fa0d13388cbb0587b","subject":"Merge \"Do not use glance_store 0.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4915fb8793a89374fb5a881fa0d13388cbb0587b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-07 19:05:08.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-07 19:08:18.000000000","tz":60},"subject":"add hw_pmu metadef","message":"add hw_pmu metadef\n\nThis change adds a new hw_pmu metadef property to the\nOS::Compute::LibvirtImage namespace and extends the train\nmetadef changes releasenote.\n\nDepends-on: https://review.opendev.org/#/c/671338/\nChange-Id: I2f4e7ef4f816afbd0524a6265d5605afa1da5f86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcb92b9978d80983ca3df4f0eada176ec554de82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcb92b9978d80983ca3df4f0eada176ec554de82"}]},"branch":"refs/heads/master"},"05493c3210b762669820015a601d02669f8d4dae":{"kind":"REWORK","_number":2,"created":"2019-08-07 20:21:26.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/82/675182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/675182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/675182/2"}}},"commit":{"parents":[{"commit":"4915fb8793a89374fb5a881fa0d13388cbb0587b","subject":"Merge \"Do not use glance_store 0.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4915fb8793a89374fb5a881fa0d13388cbb0587b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-07 19:05:08.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-07 20:21:00.000000000","tz":60},"subject":"add hw_pmu metadef","message":"add hw_pmu metadef\n\nThis change adds a new hw_pmu metadef property to the\nOS::Compute::LibvirtImage namespace and extends the train\nmetadef changes releasenote.\n\nDepends-on: https://review.opendev.org/#/c/671338/\nChange-Id: I2f4e7ef4f816afbd0524a6265d5605afa1da5f86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05493c3210b762669820015a601d02669f8d4dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05493c3210b762669820015a601d02669f8d4dae"}]},"branch":"refs/heads/master"},"2998c795ed5f1accf041373c0b4cbf0883357862":{"kind":"REWORK","_number":3,"created":"2019-08-21 00:06:44.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/82/675182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/675182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/675182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/675182/3"}}},"commit":{"parents":[{"commit":"4915fb8793a89374fb5a881fa0d13388cbb0587b","subject":"Merge \"Do not use glance_store 0.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4915fb8793a89374fb5a881fa0d13388cbb0587b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-07 19:05:08.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-21 00:06:04.000000000","tz":60},"subject":"add hw_pmu metadef","message":"add hw_pmu metadef\n\nThis change adds a new hw_pmu metadef property to the\nOS::Compute::LibvirtImage namespace and extends the train\nmetadef changes releasenote.\n\nDepends-on: https://review.opendev.org/#/c/671338/\nChange-Id: I2f4e7ef4f816afbd0524a6265d5605afa1da5f86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2998c795ed5f1accf041373c0b4cbf0883357862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2998c795ed5f1accf041373c0b4cbf0883357862"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
