)]}'
{"id":"openstack%2Fglance~675512","triplet_id":"openstack%2Fglance~master~Ia6a1c4ae557853dce7ac4cfa1adc3df774121121","project":"openstack/glance","branch":"master","topic":"fix-url-404","hashtags":[],"change_id":"Ia6a1c4ae557853dce7ac4cfa1adc3df774121121","subject":"Updating Ceph 404 URLs","status":"MERGED","created":"2019-08-09 03:47:28.000000000","updated":"2019-08-20 15:21:12.000000000","submitted":"2019-08-20 15:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675512-1566314360213-ff059232","meta_rev_id":"b285f339178cc53efab9cf59fae4834b21142a0c","_number":675512,"virtual_id_number":675512,"owner":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-08-20 15:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-19 13:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-20 14:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-08-20 14:04:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 13:33:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-20 14:04:26.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-08-20 15:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b0921be7cb1bd6d54a12b14ea9449980316d91c","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-09 03:47:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"912628d5b36125aee7e877abaa408b94c4a5c439","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-09 03:47:41.000000000","message":"Topic set to fix-url-404","accounts_in_message":[],"_revision_number":1},{"id":"5aabdfa532171eef81a6f267ea84308d54f3676b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 03:54:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/12/675512/1/check/openstack-tox-pep8/cc76e49/ : SUCCESS in 4m 42s\n- openstack-tox-docs https://logs.opendev.org/12/675512/1/check/openstack-tox-docs/e4b4a52/html/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"220493bfa0fb668757535fcf8232df4be736b489","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-19 13:33:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b67d7c65efa3fdaae1c75beb67f777265a78f5d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-08-20 14:04:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"29e4acd2580672a78adfae1cb91ea5824aaded01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:04:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"810298b2c1e9af29f3d1d1a4fe43a53328f2a452","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-20 14:30:04.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3c8634355c485108573f2bfbb46ab56826a9a063","accounts_in_message":[],"_revision_number":1},{"id":"9e669b0abeda4ec2ba8fe94e5680e0699f246b32","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-20 14:30:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 42007318925f9c677fcbabf4b1fa224cddbce3af","accounts_in_message":[],"_revision_number":1},{"id":"80f61776a6d5e2a01bfa886ce0b306c240b930b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:19:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/313592ba74c845328e9029cbd904899d : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3bf61bdbd4b443ca9012f071a0868a9 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ca5fa6b0fbae5753e37bacc3ab8ece2f4c44568b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b285f339178cc53efab9cf59fae4834b21142a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 15:21:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd17a6947034316a44baf4a4434e297 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5283d8b54d9ed8d9e8682d4be527ede3cf2d8a8","revisions":{"f5283d8b54d9ed8d9e8682d4be527ede3cf2d8a8":{"kind":"REWORK","_number":1,"created":"2019-08-09 03:47:28.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/12/675512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/675512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/675512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/675512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/675512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/675512/1"}}},"commit":{"parents":[{"commit":"4915fb8793a89374fb5a881fa0d13388cbb0587b","subject":"Merge \"Do not use glance_store 0.29.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4915fb8793a89374fb5a881fa0d13388cbb0587b"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-09 03:46:27.000000000","tz":600},"committer":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-09 03:47:25.000000000","tz":600},"subject":"Updating Ceph 404 URLs","message":"Updating Ceph 404 URLs\n\nUpdating to latest URL. Although this is probably\nbest double checked as this was quite hard to\nfind and inconsistent since the website has moved\nand changed a lot.\n\nChange-Id: Ia6a1c4ae557853dce7ac4cfa1adc3df774121121\nbackport: stein rocky\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5283d8b54d9ed8d9e8682d4be527ede3cf2d8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5283d8b54d9ed8d9e8682d4be527ede3cf2d8a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
