)]}'
{"id":"openstack%2Fglance~677788","triplet_id":"openstack%2Fglance~master~I62159315346e99522740383dd4bb5d2cc0ee368d","project":"openstack/glance","branch":"master","topic":"bp/leverage-compression-accelerator","hashtags":[],"change_id":"I62159315346e99522740383dd4bb5d2cc0ee368d","subject":"Add \u0027compressed\u0027 option to container_format","status":"MERGED","created":"2019-08-21 16:42:21.000000000","updated":"2019-08-28 23:02:00.000000000","submitted":"2019-08-28 23:00:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677788-1567033210437-80306961","meta_rev_id":"1c1614841cf4570ce420da4cb8aaf5c5f7c7fa00","_number":677788,"virtual_id_number":677788,"owner":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},{"value":2,"date":"2019-08-28 23:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2019-08-28 12:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-08-28 05:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2019-08-28 12:58:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-08-28 05:23:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 03:18:28.000000000","updated_by":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"reviewer":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"state":"REVIEWER"},{"updated":"2019-08-28 05:23:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-08-28 12:58:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-08-28 23:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8f8b135de0f54e8594a449014e121b25a7d2415","author":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"date":"2019-08-21 16:42:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"851fe34dad66f659cf60865dcfba35c3fa5bf718","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 18:51:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0a928661789420aad6f09ed976c69ab : SUCCESS in 1h 42m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab367e3a1d8f4c5d9906543c4067394a : SUCCESS in 1h 13m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/758b5883195545a09d34cb7201c0be05 : SUCCESS in 1h 05m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6114f9e3b0f24c4aa50e409d3a277278 : SUCCESS in 1h 37m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab39022afb864134a705b72d3337ac19 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b0fba20087422c85eb3c188443f97e : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3c212301b3444c29ab2bfe19cef0717 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08e1b5d2c048402baf207cdaff67357e : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f547536d93f24fdba0f9acca022aa044 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27f36f8e9f15429c8a0aefbab1fadf5a : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4995e6822a8b4d0396e2cffd04d7722b : SUCCESS in 6m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/93c77a6dca6943f78a1ae6b98ae9370d : SUCCESS in 10m 21s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/dd845e16ae254b4b9076bc532440a852 : SUCCESS in 11m 27s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/546655657018434ca03b069ecc1a9ff6 : FAILURE in 7m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/73af94466fab4e34855c96d83ddfd498 : SUCCESS in 4m 35s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ebeaa57e76f440bb66a6af888725ee8 : SUCCESS in 1h 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9788f987462695a0f18a933dd27ae0ed45764ee2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-22 13:32:52.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"58e7895ee8b7958bfaa9616fb1c917b07d69a3ea","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-22 13:41:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3fdd8a02fe847964fdbf62b049a290daec4bdae1","author":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"date":"2019-08-26 08:07:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"886c713b8de13bd7660f9a3e626de9734c296fce","author":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"date":"2019-08-26 08:08:46.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n\nThank you for comments! Revised.","accounts_in_message":[],"_revision_number":2},{"id":"fd17edd449fd015edf976785a06c4d1cd86af914","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:44:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5667ffcb30ea492791268924e4efb315 : SUCCESS in 1h 34m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab7ef1ea0a6946079a06d17d3c0e74c1 : SUCCESS in 1h 03m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2926322b578e430faecde375fc58591b : SUCCESS in 1h 02m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9bbbb49c03e468696111974dc6932b6 : SUCCESS in 1h 18m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3860cd6522f4cdea22457c13d31b225 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a795dc2f52f46c69a9c8f4de0a5054e : FAILURE in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2217d2d353a840ba84921b28a24aa8d2 : SUCCESS in 6m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0b612f042e2b48de88d623495c408d4e : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/157cf670882540a4bc3b91bdf29d1137 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98d8288188654e559775fee7758b570b : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad1e646fed74d909d4e2ab9609440ad : SUCCESS in 8m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e7fe57573f34415cbf601a0047ef3515 : SUCCESS in 9m 19s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d65bdfb9d2884de8925e1b7c7abb196e : SUCCESS in 16m 20s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/375420c4fd8145aeabca48ff0760f1b6 : FAILURE in 6m 57s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4249e33acc264b47b327b4fdff60755d : SUCCESS in 5m 48s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d3c79b925b254a7ab7a37688b0d3d2a0 : SUCCESS in 1h 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b96fb5311c0430a4647d4b1e118e97b3899860fb","author":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"date":"2019-08-27 01:06:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3a1b013fe1f1e45b77c90c654993bc6c6ad998e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:54:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d0be31e06d1545e5be11b04f4d4410f1 : SUCCESS in 1h 42m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f7bc4c15519349cebd4750265f0a090d : SUCCESS in 1h 03m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/03d0bebd60ea47cf9ba5e441855f1742 : SUCCESS in 1h 02m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/068841291cc4480c8efe8ba3778409e3 : SUCCESS in 1h 32m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b53546cacd494d648623f7c0105324d2 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1a42a0aa7e4b5f9ef022fda6e83d9c : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46aedceecff9452db19d6acb2a319250 : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b52a30bf30d04c7590913f0e89e01e27 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f8ef82e59b4c1bb75f0429ed72fa99 : FAILURE in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab3739b811a94603a61d2417ab16838d : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4b6e5d95984f71925e264a2337f22d : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1b91251dfe9943b097ffe384c4b3be96 : SUCCESS in 8m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/33adb24fcba94476939a53d23ea9dbab : SUCCESS in 12m 36s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/250c3a448c4446a6bd33da4f95fe6365 : FAILURE in 7m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4a8e5662fa214fc5a7acdbd4637e9435 : SUCCESS in 5m 46s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/674edf0adbf744508337bd11a71d2a72 : SUCCESS in 1h 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fcbe191543b7a87e1c1766fce57b0052f493778d","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-08-27 03:18:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e1ce3177bb38fa5869a522dae1794a687734db65","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-08-27 03:19:49.000000000","message":"Patch Set 3:\n\nrechek\n\nIt would be nice to update doc in [1]\n[1] https://docs.openstack.org/image-guide/image-formats.html","accounts_in_message":[],"_revision_number":3},{"id":"c29e751fad79c57ffb29242c933aa2ee309c0219","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-27 14:15:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nSome revisions noted inline.\n\nAlso, I put up a release note patch for this: https://review.opendev.org/#/c/678835/\n\n@Dongcan: Good catch about the user-manual docs.  I put up a patch to address it: https://review.opendev.org/678856","accounts_in_message":[],"_revision_number":3},{"id":"355ddab9bdfa07454832c67597e61d7c27628fd4","author":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"date":"2019-08-28 03:30:04.000000000","message":"Patch Set 3:\n\n\u003e (2 comments)\n \u003e \n \u003e Some revisions noted inline.\n \u003e \n \u003e Also, I put up a release note patch for this: https://review.opendev.org/#/c/678835/\n \u003e \n \u003e @Dongcan: Good catch about the user-manual docs.  I put up a patch\n \u003e to address it: https://review.opendev.org/678856\n\nThanks Rosmaita for these lines and Dongcan\u0027s comments. Doc has been revised. I think words are good enough so I almost copied them..","accounts_in_message":[],"_revision_number":3},{"id":"e3517fed4d14600f649839986944b89fd58a9a5c","author":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"date":"2019-08-28 03:30:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba0b1cff4585cd35ede82755037e4783f01933fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 05:17:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e4e6fc82d2104dd7b3b439af07319b45 : SUCCESS in 1h 41m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2ea245873afb489cb89bb93e0eb327ed : SUCCESS in 57m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/839d65d5c8144d9cb55a1362a3b27a3b : SUCCESS in 1h 00m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/030c976ed51b49e9b8fff3bc0d1b4d78 : SUCCESS in 1h 35m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0fee072a09746cf8cf37880e80799ba : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b444d0092c9b42ae97fef581719e385a : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1cf851abfc24b32954dd93625ed1be3 : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ef8ffb126e44e348c4b025054b2fbc0 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8158681ef484260a047d41b7ee7638a : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c727a260185347609ed59ca85bc6130d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de80b3d925fa4f06bb419861a0c7a90d : SUCCESS in 6m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/79f30ecc740e49a19d41ad11ba603602 : SUCCESS in 8m 52s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fb2da80de35248209f5f282085588b1e : SUCCESS in 9m 17s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/277a4e3cad5c4a95a4db6eb6a2542426 : FAILURE in 4m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/31f74d7a9041403f93bbe880a2d7196e : SUCCESS in 4m 26s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13746520cad24028955bc5a9c99950b8 : SUCCESS in 1h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4be3ecb9352c23e3c10bcb472a3152bc2c83fe7e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-08-28 05:23:07.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"59ceb9c41804e0cd13a0a5373f79f4e0242dd91c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-28 12:58:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nAll my concerns have been addressed.  Looks good!","accounts_in_message":[],"_revision_number":4},{"id":"721d063b787feeea839dd629e97ba920a08c4859","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 12:58:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"74bdae6b52aa012709d94f504538451e70a6c6db","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-28 14:06:15.000000000","message":"Patch Set 4:\n\nopenstack-tox-functional-py35 has failed because of the subunit parser problem -- go ahead and recheck when zuul finishes the tempest tests.","accounts_in_message":[],"_revision_number":4},{"id":"e080d84a8cf3ba006b22d545085f4daca85d8df2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 14:33:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/189a6ee296104ad5b1575c022345ea73 : SUCCESS in 1h 32m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2372ad17cf864c5eba4cc2826710434d : SUCCESS in 52m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2809561c08504c13b4b8e583a393725a : SUCCESS in 1h 03m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8886bd11eced4dd080d98ac9c016ce16 : SUCCESS in 1h 25m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fae910377b694ca98e2a9e07434eb53b : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf2826c8491483b912c9f97c2d94106 : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/423ad07125264a869f0e7d4a66d61ec9 : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5deb44a8d209496eb0328fd129dcaccf : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5391715c0830455aacaf8c435482781d : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f3f7f1b99bb4824b9bc751402050340 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccae00a0121d430f9ab4acbf7023b4b1 : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/24d7e96b48134e158c7fa2006c3cc8a6 : SUCCESS in 8m 54s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/a110f1276902467ca5b98ad94fb2ecd7 : FAILURE in 10m 47s","accounts_in_message":[],"_revision_number":4},{"id":"bc22f79822e37a60c5934f36fd192d2370a90cbb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-08-28 14:50:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"879ca1ae2b4dde005a7f605925e4e864676e6f56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:16:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/29500ebec60b42aeb242cedd25d8c4ac : SUCCESS in 1h 51m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0748ddc8c9cb44f2b850de573d2c7e6f : SUCCESS in 1h 11m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c75cacf4f6dc4a9eb7a29be15b96b925 : SUCCESS in 58m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48cb08ae4c7f4d068c9e6753498690f9 : SUCCESS in 1h 36m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a76c4c33feb4afe8f773deba83d9c3d : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5171d4244f749298de56dc943914b73 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c0656aaf8524d67a6f0cb47703c1850 : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be70eb79bab945fd8b26d3b76197cb7d : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60f2711634f8407fa4966bcffd3d6821 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b10d6cbbff24055beabec56c94f1824 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/655d4b7687bc409a840567a9463064d2 : SUCCESS in 6m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/88658a7e62034d24b575e58421242b81 : SUCCESS in 10m 18s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/76a70f758c8b4be5b1ef2004a041e2da : SUCCESS in 10m 42s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/9759c405994b4520a38b0b8db93d2777 : FAILURE in 5m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f587bb3eafd04bbb820846fc2319390b : SUCCESS in 4m 27s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d1783b6a4bb4ab790fbea7d9f1445c9 : SUCCESS in 1h 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ce3046823636aa20c422b8c47101c8b3e6620494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 18:17:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c9c5ece74e2f617283bf945d353eb00f3d18ecc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 23:00:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/522f7d89a1bf4b5fb9a7d34824669472 : SUCCESS in 1h 53m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1460ab9d007e46d5b8d282ac5bbeb10b : SUCCESS in 1h 01m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d3f9284d054a4feb89ad15a15d2b6575 : SUCCESS in 1h 02m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0e4a7bb5ec744b289a5c5ad41467b5f : SUCCESS in 1h 17m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1f5ef30133349cfac643ba482fdb2f4 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ef0244b6694ecf9b4bfa8c486fe573 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19f28264f0bd43e3ae8fc0bd71f5321a : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/45beb6ab59894264a880aa1eee87b84d : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a2bcaee0c5742239aa0a90387ba19f6 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99dacf4d77234683bc4b3f7d76e3de5e : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d451944ec247b0ab89ecadd127f896 : SUCCESS in 7m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b0244fb763944952b8c11e77972250b6 : SUCCESS in 9m 23s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/831c3f5f49b04bac816a881371262a56 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2e98d85ebb5ca27248d9755b1675c8c679e9b227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 23:00:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b58548d6e5fd8566535dcd395d7daf0cbefee768","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 23:02:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0769bc74fec413e80da79d95d8335c9 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"43aa0479229c23c828530f25dac5208395761b5f","revisions":{"13a1fcf603dd047959ae76d988659adbda13ba54":{"kind":"REWORK","_number":1,"created":"2019-08-21 16:42:21.000000000","uploader":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"ref":"refs/changes/88/677788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/677788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/677788/1"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-21 16:31:53.000000000","tz":0},"committer":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-21 16:31:53.000000000","tz":0},"subject":"Add \u0027compressed\u0027 option to container_format","message":"Add \u0027compressed\u0027 option to container_format\n\nThis patch is purposed to support a new container_format\nso that we can do image compression when uploading a volume\nto glance and decompression when downloading a image from\nglance.\n\nThis patch includes:\n\n1. A new container_format option: \u0027compressed\u0027.\n\n2. Unit test for new option.\n\nDepends-On: https://review.opendev.org/#/c/670454/\nChange-Id: I62159315346e99522740383dd4bb5d2cc0ee368d\nImplements: blueprint leverage-compression-accelerator\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13a1fcf603dd047959ae76d988659adbda13ba54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13a1fcf603dd047959ae76d988659adbda13ba54"}]},"branch":"refs/heads/master"},"b9dcd9f4493b16a008d0b66a4c8e299f9acd837e":{"kind":"REWORK","_number":2,"created":"2019-08-26 08:07:49.000000000","uploader":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"ref":"refs/changes/88/677788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/677788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/677788/2"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-21 16:31:53.000000000","tz":0},"committer":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-26 08:03:04.000000000","tz":0},"subject":"Add \u0027compressed\u0027 option to container_format","message":"Add \u0027compressed\u0027 option to container_format\n\nThis patch is purposed to support a new container_format\nso that we can do image compression when uploading a volume\nto glance and decompression when downloading a image from\nglance.\n\nThis patch includes:\n\n1. A new container_format option: \u0027compressed\u0027.\n\n2. Unit test for new option.\n\nDocImpact\nImplements: blueprint leverage-compression-accelerator\n\nDepends-On: https://review.opendev.org/#/c/670454/\nChange-Id: I62159315346e99522740383dd4bb5d2cc0ee368d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9dcd9f4493b16a008d0b66a4c8e299f9acd837e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9dcd9f4493b16a008d0b66a4c8e299f9acd837e"}]},"branch":"refs/heads/master"},"e57e78840ef111884d5006722476949ab723df4b":{"kind":"REWORK","_number":3,"created":"2019-08-27 01:06:44.000000000","uploader":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"ref":"refs/changes/88/677788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/677788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/677788/3"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-21 16:31:53.000000000","tz":0},"committer":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-27 01:06:27.000000000","tz":0},"subject":"Add \u0027compressed\u0027 option to container_format","message":"Add \u0027compressed\u0027 option to container_format\n\nThis patch is purposed to support a new container_format\nso that we can do image compression when uploading a volume\nto glance and decompression when downloading a image from\nglance.\n\nThis patch includes:\n\n1. A new container_format option: \u0027compressed\u0027.\n\n2. Unit test for new option.\n\nDocImpact\nImplements: blueprint leverage-compression-accelerator\n\nDepends-On: https://review.opendev.org/#/c/670454/\nChange-Id: I62159315346e99522740383dd4bb5d2cc0ee368d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e57e78840ef111884d5006722476949ab723df4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e57e78840ef111884d5006722476949ab723df4b"}]},"branch":"refs/heads/master"},"43aa0479229c23c828530f25dac5208395761b5f":{"kind":"REWORK","_number":4,"created":"2019-08-28 03:30:11.000000000","uploader":{"_account_id":30648,"name":"ZhengMa","email":"zheng.ma@intel.com","username":"ZhengMa"},"ref":"refs/changes/88/677788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/677788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/677788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/677788/4"}}},"commit":{"parents":[{"commit":"e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a","subject":"Merge \"Windows: fix flaky tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e593e9bc866e9be5842ab6a8ddfe1f24875c7d0a"}]}],"author":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-21 16:31:53.000000000","tz":0},"committer":{"name":"ZhengMa","email":"zheng.ma@intel.com","date":"2019-08-28 03:01:48.000000000","tz":0},"subject":"Add \u0027compressed\u0027 option to container_format","message":"Add \u0027compressed\u0027 option to container_format\n\nThis patch is purposed to support a new container_format\nso that we can do image compression when uploading a volume\nto glance and decompression when downloading a image from\nglance.\n\nThis patch includes:\n\n1. A new container_format option: \u0027compressed\u0027.\n\n2. Unit test for new option.\n\nDocImpact\nImplements: blueprint leverage-compression-accelerator\n\nDepends-On: https://review.opendev.org/#/c/670454/\nChange-Id: I62159315346e99522740383dd4bb5d2cc0ee368d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43aa0479229c23c828530f25dac5208395761b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43aa0479229c23c828530f25dac5208395761b5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
