)]}'
{"id":"openstack%2Fglance~681784","triplet_id":"openstack%2Fglance~master~I13133c32734751d43cc3afed9b68d015cebd5b6b","project":"openstack/glance","branch":"master","topic":"api29-n-configs","hashtags":[],"change_id":"I13133c32734751d43cc3afed9b68d015cebd5b6b","subject":"Images API version bump and config sync","status":"MERGED","created":"2019-09-12 15:35:07.000000000","updated":"2019-09-17 04:38:06.000000000","submitted":"2019-09-17 04:36:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681784-1568694985265-c3ff88a4","meta_rev_id":"7d8d6e9d4bdfd47fde9a159229071d00a5c1f781","_number":681784,"virtual_id_number":681784,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-16 19:36:03.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-16 16:33:42.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-09-17 04:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-09-17 01:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-09-16 16:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-09-17 01:47:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-09-16 12:26:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 07:20:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-09-16 19:36:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-17 01:47:38.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-17 04:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d63934ce57529021c8e1b413cbc966518368ff1","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-12 15:35:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc23ba19a4ff45646ec40a0bbda32bc08309bea8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:27:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a0df0e68ab35484a9f0425318479311a : SUCCESS in 3m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c38f68791b84c2a905b85a2d63ffad8 : SUCCESS in 1h 40m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/924ed2adf4ca4b629826f450f107def1 : SUCCESS in 1h 03m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ff3d18db29744523ab32fea0f3666acd : SUCCESS in 1h 04m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79afcbc5233548298915897528eb0720 : SUCCESS in 1h 25m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a705c7f26a6b4e47aa1512e2cebe68d9 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e85beb0cd9c44faa3a25d20b26de9fa : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e3fc830324049118769237a6c18c91b : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/342dbd039a47455d897404c7d46ba48e : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84d6beeefe384d54bfc8277ce8a4f102 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8fc95235fb45ecab900c5f4ff4d712 : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5d0c57d5ef347d7baeb7a273787238b : FAILURE in 9m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4b1cb8b13e81488fb94d21b4856e28de : FAILURE in 22m 49s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/7fd1a5dbb7684e9eb650318f1279fded : FAILURE in 4m 37s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f18c155b750486e847b5ff237ac5f9b : SUCCESS in 3m 24s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19b6807430ca49ebb1134239d983c6cf : SUCCESS in 1h 06m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70956517c68c8cf9573ea21f6ee00df1e2afbf58","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-13 03:57:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b20c521e9ff4912c78916c6fbf53bd79855ffc1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-13 07:20:28.000000000","message":"Patch Set 2: Review-Priority+2\n\nNeed to regenerate as https://review.opendev.org/#/c/674697/ is merged.\n\nAlso we need to get https://review.opendev.org/#/c/611253/19 in before M3 as above config change patch is merged.","accounts_in_message":[],"_revision_number":2},{"id":"d9fcccddc9d59a1c88171df981412918d776c9c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:03:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6631b2499c9480cbb6c7e65f7f93913 : SUCCESS in 4m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27fd1e8bbe4340798cf3cd2b9c6ca8df : SUCCESS in 1h 37m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/24a0aad67cf2416cbc2f8fba8f9a48c3 : SUCCESS in 54m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d4b74558166f4cd791e850266f920d07 : SUCCESS in 1h 03m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ec31ed5713d4786a8f622bf943188ce : SUCCESS in 1h 27m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/143088122b17449eb15c9b9bce9cddff : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/736d7f94fa3b4a8b8746a7a4db0d4be6 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/581b07a47c834738847832b7e60bb05a : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f32b6956f60b406c8aeab2857aa5d4c4 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fee1ac02f5044d4864330354c50ab4f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1cc266ebfec47cb971358fcace2469a : SUCCESS in 6m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e7b6835c8cd14faa81a228958a112925 : SUCCESS in 9m 52s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0de3370edf7d4b3c91a853cbba740ec4 : SUCCESS in 13m 14s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/7b4a355065e44d669fd5743abd1bb668 : FAILURE in 4m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9b059f8346d5476396063e93f270c14f : SUCCESS in 6m 05s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6babde48b79d4dfbac226276dbb4233b : FAILURE in 1h 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a17b8ca1bb34ffffbb8de4185c32f74269c8c50a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 11:20:13.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)\n\nThis is out of date.  HEAD in master is currently 99f3235a.  This is missing the config changes from ff7d940b.  This can\u0027t be fixed in gerrit, it has to be rebased locally and the config regenerated.\n\nAlso, comment inline about the versioning.","accounts_in_message":[],"_revision_number":2},{"id":"03530e3b91aceeb05a28d353e2b355cc20a6118f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-13 12:17:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf6321548847e9b39da3e22d40ee66a8dae442d0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-13 12:28:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d499362c5e71593f1d0774c9030dd9aa6bbf0bf0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 12:39:48.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nResponse inline.","accounts_in_message":[],"_revision_number":2},{"id":"ed187a1c89394c493f46d512d545b8712e279642","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 12:42:32.000000000","message":"Patch Set 3: -Code-Review\n\nSpot checked, looks like the config files are up to date.","accounts_in_message":[],"_revision_number":3},{"id":"6e7b519aa277691a5c85eb5869dd234d2f7e4db6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 13:13:48.000000000","message":"Patch Set 3:\n\nThe config files will need to be regenerated after https://review.opendev.org/#/c/682041/ merges.","accounts_in_message":[],"_revision_number":3},{"id":"0f2564822edeb542f457cd2d68e7c05bea2c2d69","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-13 18:40:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e87a3021081e198801e8b362fb430ce81060976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 23:29:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c7a9f4e221db4381955ab8f13a16d2b9 : SUCCESS in 5m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f0344ef98de9478b972d4ad9c67562b7 : SUCCESS in 1h 57m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4fa46fbb6a4f40629a6ee005fbb8ecf9 : SUCCESS in 1h 01m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9374e421e4f948308a3af5cbb729dea0 : SUCCESS in 1h 06m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99a5d47b91ed46d792c6d5940612171d : SUCCESS in 1h 31m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3fd1305399e4c1d86abe20c0ac7113b : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833b204943f247e79e71a4225b2df11a : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c251628f3a044ce49b84eab5f59ad8cd : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79179477cad443f6bf2e8f549d9a0335 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05cba320513e49c6968fc380650903f3 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dffced3a6c64cfa818f5fcb371a9f99 : SUCCESS in 6m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3712c1c57e4d4eea919ad3532973cbf3 : SUCCESS in 11m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0210d50dc06146448ada90c9902bf110 : SUCCESS in 9m 12s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/3a8e14ea4ea34ff7a3be804c23c39880 : FAILURE in 4m 03s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/697a18d74f194f408f71b53b96d6bc71 : SUCCESS in 4m 21s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54afb2d0ad974282ac9eb8d4071a1dfb : SUCCESS in 1h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41aa97a7352360d98fbf620c0a9c85c5e1c5ac1b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-14 04:23:41.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/84/681784/3/check-not-vote/ext-glance-zuul/f6eb75a : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":3},{"id":"8d75fcd1abe881456cde2ff29d78357883d08fab","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-16 12:26:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e51b0c61591a9e76be776b1bdea9f2c7a326576e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-16 13:13:54.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/84/681784/4/check-not-vote/ext-glance-zuul/3fc94a9 : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":4},{"id":"288baacd009c67d3d9dabd05ab67c2c7e5c041d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 16:18:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b5eeae9e2d2444faee102d617ec27ba : SUCCESS in 5m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c7fbf8529af408c9f5b62448a550349 : SUCCESS in 1h 47m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/06369478b9a04841a48e0dc11e2ef6a8 : FAILURE in 44m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/76c85451782b49dc807c978631220760 : SUCCESS in 1h 05m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/029cc558ff7647feb1f474992f78d540 : SUCCESS in 1h 31m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d52d57d915354bc8a259aabbd83fceee : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2156ac98d40748c9bf5feaa37c75f460 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b6335e4013c442ca90af53f1a8cf2d6 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/390d9234ab504d8babc619e226bf95b4 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7c5f319057b4355b3f14cd1437ae806 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8516074966d44562a74f27d0092b2aac : SUCCESS in 8m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7b9b4831e9c04b6a88ffa2bcec04daac : SUCCESS in 10m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/98a71762ab674c7dac810977a751a01a : SUCCESS in 9m 54s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/48d4029b2ce740d29afe807045606652 : FAILURE in 5m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3790cd6a5f754da59b523965bf2c5e6c : SUCCESS in 4m 52s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e69b194cbdf44db08770fc5f479e2e71 : SUCCESS in 1h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"352599c379b84e7b8e6abec0fb66bfac99111b8f","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-16 16:33:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e0062130380644de5278b9d2102b29c6c247a195","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-16 16:42:46.000000000","message":"Patch Set 4: Code-Review+2\n\nFollowed the discussion on IRC channel, satisfy with the change.\nThank you erno!!!","accounts_in_message":[],"_revision_number":4},{"id":"db300599dd81532d627358f3bea928e06282ac89","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-16 19:36:03.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/84/681784/4/check-not-vote/ext-glance-zuul/dbcd3f4 : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":4},{"id":"15025271177861bece1fe985ae2ea5437ab041fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 00:13:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7439dab71efa4b678918a283231be571 : SUCCESS in 5m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f66ff664534c4a1da8cb3845c9e77c65 : SUCCESS in 1h 34m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/383ee436272f46b195f37a3e3c462893 : SUCCESS in 1h 09m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/845e836051384ec9bb2df1c18b7bdf18 : SUCCESS in 1h 12m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f750cdf56a045cc96ba327883e285b1 : SUCCESS in 1h 22m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8daa177bf78044cf91d5747eb6dfae31 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298922330d904ee99209be514de91329 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/987f7967547246c0b8ab8f67b44930d4 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88aa05efd0af4cd793dc043567da5e07 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/640741e95db64bc78cee3601928f67e3 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2fadb535d7471eb66161db0fa50a6a : SUCCESS in 6m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1240d84f5d10408d9fabc71fe9b563b8 : SUCCESS in 8m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ab2906d8de27496cbef9861eaad3786a : SUCCESS in 10m 44s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/e2b2f1ee05074223a578054a7f530ac0 : FAILURE in 4m 47s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ec39551902954b50a556768a3af03211 : SUCCESS in 4m 08s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/990d9b61417b4c7088899fb0da536bd4 : FAILURE in 1h 55m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6219981e6c4e3e9f01a4c79d1dfa65885d78e3b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-09-17 01:47:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"88815e9ba4fa61a9f8a273584963e89202cb1467","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 01:47:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"28047cfd6dd0a267f90e0c59fea9cc7bff4f6648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 04:36:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3f865ac527b84858b2e704e4a592793b : SUCCESS in 3m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4c3c3f9c68274f849df9ff366477c6dc : SUCCESS in 1h 47m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6cfb9178b32d4fc18fb0f5a954391fc9 : SUCCESS in 1h 10m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1a767b84d1dc47d58254d7d5abc6ad9e : SUCCESS in 1h 04m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15d0eba569d248f6b074045c61c2ebb4 : SUCCESS in 1h 31m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/297858e98db94e1b8fc9af12bf4b8c64 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d9f96196b414015b3ba58a58babfda9 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8fd3a1855e34e8ebdf14b3346c7a494 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77266e1d6b3a4b29b57fbb13e2d2be1a : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/292913b2193f441596e681d6af88880f : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccfb3a0bec2d486ab29c0f33e2c3e68c : SUCCESS in 6m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7bff9d9992ff40b1b96335dd31bc0d5c : SUCCESS in 7m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ab32f9e53bb844b59f2c747c04589646 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":4},{"id":"87e5caa60a4a2ea133df517de56c782ada1b2e3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 04:36:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3a66b1364e7ce2b9cb33e399cd344c0fc68cef33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 04:38:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9d971ee2820b47729bcb7a188589656c : SUCCESS in 1m 10s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5337b74c2c4f14901cfc271a7c3977 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d9546dce754c68084944ec01aaabd33790f8990a","revisions":{"121019d771187da8213bbbb933476b0c0c3150fd":{"kind":"REWORK","_number":1,"created":"2019-09-12 15:35:07.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/84/681784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/681784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/681784/1"}}},"commit":{"parents":[{"commit":"73fefddd969e75e2f2d82866f4b8b9eeca488770","subject":"Add periodic job to prefetch images into cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73fefddd969e75e2f2d82866f4b8b9eeca488770"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-12 15:31:33.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-12 15:31:33.000000000","tz":60},"subject":"Images API version bump and config sync","message":"Images API version bump and config sync\n\nThis change contains minor version bump to Images API to\nindicate latest changes to the API.\n\nAlso has config file sync for for M3 release.\n\nThese two are combined for saving some gating time due to\nbusy infra at Train milestone 3 deadline.\n\nChange-Id: I13133c32734751d43cc3afed9b68d015cebd5b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/121019d771187da8213bbbb933476b0c0c3150fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/121019d771187da8213bbbb933476b0c0c3150fd"}]},"branch":"refs/heads/master"},"44d851f027b8f84ae796b982f01c4a18c4386179":{"kind":"REWORK","_number":2,"created":"2019-09-13 03:57:05.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/84/681784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/681784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/681784/2"}}},"commit":{"parents":[{"commit":"73fefddd969e75e2f2d82866f4b8b9eeca488770","subject":"Add periodic job to prefetch images into cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73fefddd969e75e2f2d82866f4b8b9eeca488770"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-12 15:31:33.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-13 03:56:51.000000000","tz":60},"subject":"Images API version bump and config sync","message":"Images API version bump and config sync\n\nThis change contains minor version bump to Images API to\nindicate latest changes to the API.\n\nAlso has config file sync for for M3 release.\n\nThese two are combined for saving some gating time due to\nbusy infra at Train milestone 3 deadline.\n\nChange-Id: I13133c32734751d43cc3afed9b68d015cebd5b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44d851f027b8f84ae796b982f01c4a18c4386179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44d851f027b8f84ae796b982f01c4a18c4386179"}]},"branch":"refs/heads/master"},"666228df28023170269fe1adf1621f7aa08760f5":{"kind":"REWORK","_number":3,"created":"2019-09-13 12:28:14.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/84/681784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/681784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/681784/3"}}},"commit":{"parents":[{"commit":"99f3235acf39a9be534d60d3cbcb338fcf15bf07","subject":"Merge \"Multiple backend support for scrubber\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99f3235acf39a9be534d60d3cbcb338fcf15bf07"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-12 15:31:33.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-13 12:27:59.000000000","tz":60},"subject":"Images API version bump and config sync","message":"Images API version bump and config sync\n\nThis change contains minor version bump to Images API to\nindicate latest changes to the API.\n\nAlso has config file sync for for M3 release.\n\nThese two are combined for saving some gating time due to\nbusy infra at Train milestone 3 deadline.\n\nChange-Id: I13133c32734751d43cc3afed9b68d015cebd5b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/666228df28023170269fe1adf1621f7aa08760f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/666228df28023170269fe1adf1621f7aa08760f5"}]},"branch":"refs/heads/master"},"d9546dce754c68084944ec01aaabd33790f8990a":{"kind":"REWORK","_number":4,"created":"2019-09-16 12:26:00.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/84/681784/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/681784/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/681784/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/681784/4"}}},"commit":{"parents":[{"commit":"b639ed080d5b74fcc3081a09bd64dd4703dea5a4","subject":"Correct the deprecation messages of local dir config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b639ed080d5b74fcc3081a09bd64dd4703dea5a4"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-12 15:31:33.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-16 12:25:45.000000000","tz":60},"subject":"Images API version bump and config sync","message":"Images API version bump and config sync\n\nThis change contains minor version bump to Images API to\nindicate latest changes to the API.\n\nAlso has config file sync for for M3 release.\n\nThese two are combined for saving some gating time due to\nbusy infra at Train milestone 3 deadline.\n\nChange-Id: I13133c32734751d43cc3afed9b68d015cebd5b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9546dce754c68084944ec01aaabd33790f8990a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9546dce754c68084944ec01aaabd33790f8990a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
