)]}'
{"id":"openstack%2Fglance~681787","triplet_id":"openstack%2Fglance~master~Ib47e96f23277cacbe7b8b9bc9907ea14aca69d75","project":"openstack/glance","branch":"master","topic":"bug/1843772","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-07-01 16:15:47.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-01 16:15:47.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-01 13:43:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"2062":{"account":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"last_update":"2026-07-01 16:15:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib47e96f23277cacbe7b8b9bc9907ea14aca69d75","subject":"Make glance-manage honor config-dirs","status":"MERGED","created":"2019-09-12 15:46:08.000000000","updated":"2026-07-01 16:17:03.000000000","submitted":"2026-07-01 16:15:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681787-bug/1843772","meta_rev_id":"073df97c79af7f3c196a0f9459eb8ce60ac7805d","_number":681787,"virtual_id_number":681787,"owner":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-01 16:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-03-12 07:29:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-03-01 06:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-01-30 01:22:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-01-30 08:10:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-03-29 07:37:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2026-07-01 13:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2026-07-01 13:43:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 17:53:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-18 01:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-23 07:17:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-23 07:17:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-01-29 15:45:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-01-30 01:22:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-30 08:10:33.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-03-12 07:29:11.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-03-29 07:37:30.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-04-02 03:03:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-04-02 03:03:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"dce98e8f7818cf9aae998945a8690e380c07efc0","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-12 15:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee09189392a975b05f3f33daf7950387b3a8f970","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 03:36:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2c94fefc133641f89d623c359a43b4b3 : SUCCESS in 1h 45m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/60888b00c3734fd18f177b1c5b012907 : FAILURE in 59m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/83345c8ba3d444d1883f75d1781dc407 : SUCCESS in 1h 04m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/069649e5c60c481fa66fc4806c207a5d : SUCCESS in 1h 27m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3b72309f5b84fa3978827fd6d676506 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b671f13a0f6b4326b191bc365b275f71 : SUCCESS in 7m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f450f83422542b1b9edf16cfb7537fe : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b689004ca9454d22b2e70558e926274f : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/889ee17b46fd4adba2de5b1cd3a0a0c4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e54f4c70ca41109358725662b28773 : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f5c896ea35f94476b26aec756a0d6f39 : SUCCESS in 9m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/35595de272884d0285682734c27ae7e5 : SUCCESS in 10m 30s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/f1d4105a6dd04ea3b8d9be665c1f667c : FAILURE in 4m 06s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d8cc2704bc794cb7bcc6af3c8dc639e7 : SUCCESS in 5m 52s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/01cfa857130a4c2b86614fa2e03d2e30 : SUCCESS in 1h 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92ab8c088fffbdbbc5b871deab18f93dfbe32359","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-13 05:17:18.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/87/681787/1/check-not-vote/ext-glance-zuul/28c12e7 : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":1},{"id":"244fc5a4bcd9e0dff5c5b07de7b18521a341eec3","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-13 07:33:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d28e63e8f7b13f02501ab3b73c9405e3ef02b9a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:30:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/42511a40fbf94a1f96a3e89e3e4fbd1a : TIMED_OUT in 2h 06m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4f20f39095f84f399aadc6812c94ca99 : SUCCESS in 1h 02m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/880418e903a540d9aff0e50c4d67b435 : SUCCESS in 1h 05m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4f9636bedab4834a5bb5dee2e133922 : SUCCESS in 1h 42m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3641fddd3fef4dfc8ec40b9feb5d8956 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7800c6ddd6dc435ca2caded6ef5648d1 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24403a4afdfe408c86a4c80938810606 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfba71cb63584e0d92147d1a6b799609 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c30f0ba7500469f97574a3c3c70859c : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0391fc650fa446abf91634c6c0ce030 : SUCCESS in 6m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c0ddfc570224a1885732384f28097ae : SUCCESS in 8m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/87b4c1c1b1574b4e843c1fbdbec48a03 : SUCCESS in 8m 09s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/f08d7fea09064173ae90901f06698146 : FAILURE in 3m 55s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2c8e73e572da47abb7801243326ca4aa : SUCCESS in 3m 49s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a98868d763ad462c8eef15355775cca0 : SUCCESS in 1h 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aecdf1ee41c31a88d5931997a0177393dd9a6412","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-13 18:07:46.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/87/681787/1/check-not-vote/ext-glance-zuul/431f485 : FAILURE in 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"35239ff201994972b6511fea0064a19d063fc37f","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-16 07:36:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb51b086474a52c1a3f39d25ea544d4798553415","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 10:14:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/94dbd5eda87e48b3b11300cb10e1249a : SUCCESS in 1h 39m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/33968e9573934078b6be86356b9dba3a : SUCCESS in 1h 13m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/11ae8325a5a34d5db2acb61adca1e40b : SUCCESS in 59m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22b808a3a66a4a5d8ad5f2a795941c8c : FAILURE in 1h 18m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5830072d2a54b7a9d85d35186e6100f : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf77f2617d7456f99b3ee81628e12ad : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10edaed2af974932b68197ff2faaf24e : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15951b81bb3d487aa98e05b3268f8ee1 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be872cd32eb40d494390d49b260527f : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44424f6fccef49cea3ea7aeccf8970e7 : SUCCESS in 7m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/abcb057857af4e3d8c5bc0154b982176 : SUCCESS in 9m 52s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9a3af3ba819a402ca09b2227e33b62e3 : SUCCESS in 9m 09s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/328868e1f805428bbb46072da965207e : FAILURE in 5m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/276e7c8fabba4652a6d20aa5e20f007c : SUCCESS in 4m 41s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cbdec56472da4e94b31f7424a8c4d3be : SUCCESS in 1h 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f07fc80f8f33bed31245063b009b190128a2d6f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-16 10:41:50.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/87/681787/1/check-not-vote/ext-glance-zuul/b985286 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e51b8a104b29f9e76613c47dcc82bdea49f0392c","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-17 15:59:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1074bd75090eb351832cf879a052b18f2bfff37a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-17 17:53:19.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/87/681787/1/check-not-vote/ext-glance-zuul/dd8ad2f : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"661753fefc840ef6975f41c28b36221c73752e16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 01:22:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/55af583d072d4dbb9c0123e7c42117b8 : SUCCESS in 1h 54m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/032d111f0e0f4e258ba7e13ff884b71b : SUCCESS in 58m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9529bf5771e14fc0a23e698b3c40feef : SUCCESS in 1h 06m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4cc6b160261446d9f39f0908c3028f8 : SUCCESS in 1h 34m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eafb9503a2a4ab48ddfca932f1c7530 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2292bd7dea8649089333589eff241d73 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef611b5345204617b2a4df75445dd5f0 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7eb2156fc434b70a76ae83f5cff0ed0 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e08ef0f968bb41a5ac56c1eb664f05a3 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d4adc39e2b42ecada80400058a2e8c : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8b475ad1facf45dcb2b4eed4f71997e2 : SUCCESS in 10m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2c4a11e6725a4c71ac632a5e87424522 : SUCCESS in 9m 20s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/0e85ff8bf0544204921fa4712ecb54cd : FAILURE in 5m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b19c79526a934859a437e59586e35daa : SUCCESS in 4m 30s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58c2941694b742b29c05f44466683e63 : SUCCESS in 1h 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26607b9897a0708a07ba8e3f3aaea276df37747f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-28 17:04:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"86974f337a71f052bae7b550e8f35b7ee64e0472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-28 17:10:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/705ea8382ab84a91be433a14cb339178\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/047bf24e16554d5c85ca8a2b55cea6af : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6a05a3275334816bb14c74bd471b0f1 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96e03fdec1134ee28ce83809588f0c35 : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"095fba654009d3aa5b0687989a0e3ba609457630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-28 19:09:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/168da69157a24dbca13a342ebb7488f2\n\n- grenade https://zuul.opendev.org/t/openstack/build/74f324fe2bb847509a9ae75d3d12d20e : SUCCESS in 58m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6f6f46f1c0b3447fb0e0ef8c60465648 : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cbac990de1d34c48a898da23428b348e : SUCCESS in 1h 34m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b5763f7606e4552b6dbe700edd6f56e : SUCCESS in 35m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52604822a5ed4c5f829770b4488caf87 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a86ed1a4d28141f583556ae95c6557d4 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0e3f35abb3242ca95ee6e2847a1e24f : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46197b484c9e49ad81f04e159b1bf2ec : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2c975a36e62d43ef85f41a4d25267de8 : SUCCESS in 13m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19168da155b24f1cbd0dca8b31e92c09 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dcdb11e926714297a8aeed59b74f32cc : SUCCESS in 9m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/db910b671fad4633a27c0f10e49d93de : SUCCESS in 8m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1cb1850efaa14d1f922115d809992f38 : FAILURE in 8m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6ee2d6ed998f42afbb6bd92758102eab : SUCCESS in 1h 50m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2632b6505f00451f8204ccaa39ee1777 : TIMED_OUT in 2h 04m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2d25bbdcbe604aa686b5b17650a8479f : SUCCESS in 1h 44m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/10f7a03ccc1b483eaac426a50bf2d741 : SUCCESS in 1h 49m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/903de6792a30445aa713c61467717730 : SUCCESS in 1h 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/041813636691469392211aecb5529840 : SUCCESS in 1h 04m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff7c726099654d869517b1eba06cf46b : SUCCESS in 1h 28m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9851641e11f49d687d8b43433fd59c0 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":2},{"id":"616a8445f25ec2479f4a1548a371ec6016810ca9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-29 15:45:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"612b649ac0c3584190c97f706c35992e08dbeaa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 16:03:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de1747c3485a4263802de08b3777b7ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e3bbcbf48e442f199d615eca7f6580c : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4941cfd8eb8546a984dd8d28864e6121 : SUCCESS in 15m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d789f002478341e9ab098371dfdf3a44 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec6b278499992ba68df817c4716fa9c245a0cea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 17:58:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2abf746ca9cd417c8f6df4d5d4dc0ccc\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ea8f09fdd6647208416f05120d4cc15 : SUCCESS in 41m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2166d46c19fc4c50a64fdd10a81d3af2 : SUCCESS in 58m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9963a012b374532802903db687332fd : SUCCESS in 1h 33m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/602b854435f74e2e8d467f6c3b5b5699 : SUCCESS in 38m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69779549741d40b296bc6e2849173141 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/447ae8840b704ba89b129a9ca1022bbe : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d9ba170cbd94d15a3dc581e121670cc : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53733310545b4821b96a5beb4e6a60c6 : SUCCESS in 9m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a5b4f62cc0ca414882b0a6394eea0094 : SUCCESS in 13m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e06f7c2c2d2d4177b30e2871631a4893 : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa6901c05d734d55b1ea8ff474ae04b4 : SUCCESS in 8m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a475fde178124fdc994b937a67541b87 : SUCCESS in 9m 08s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e74f7dc6b87b4983a0d2959df37628dd : SUCCESS in 10m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/392a41e0147b40229140f2018ca2dcce : SUCCESS in 2h 01m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/780f121733fe4ed8952477538cf7bb15 : TIMED_OUT in 2h 04m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a788ab41d0d442e782e66d97b628b9d4 : SUCCESS in 1h 37m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/10a3bc8524bb49aeaa85eda6ef7ba88e : SUCCESS in 1h 52m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ff606b43bd0a4c509ca043ae4423ec03 : SUCCESS in 1h 39m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88494bc9dc86433589e140e9207862ce : SUCCESS in 1h 10m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d19bc059e7fb4707b07b7556dbba6c55 : SUCCESS in 1h 55m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e684b9010d44227ac8a4fc2dd26d5c1 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f0e29a447b2f4b9984314fd87a7f340d426cae2c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-30 01:22:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"509fe4f14f65870a567f256e2834bebf01e511c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 01:29:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0808aa2ff9347a6a70a4ebd30f7ec36\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acb6eb257a8d4490932c5d0adf8f6a28 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/300bf4cae24b4fba8f2b65e07d96cf78 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3b25ec9dc694b8eab3bf466a7cb143e : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab048d1e9b1ccd3f2db8c3a9bd5186eab67d305e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 03:12:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa264c1056b043fc8f965fca08cfd28b\n\n- grenade https://zuul.opendev.org/t/openstack/build/79fc4918f387459194ccd22db3de0b5d : SUCCESS in 1h 08m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d0503ee217db4da286315fcf88336739 : SUCCESS in 1h 02m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc19b64f6fe346cb9e73a401e9e663c8 : SUCCESS in 1h 22m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e63f33b1532455689feb87dc56089e3 : SUCCESS in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b514813c2549be89a3d3162342b5c3 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/282d28a346e145fba8eb87cf8867a585 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bc27ea6039b4a4ab65ecde720deff92 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60b03ae6a924024b62a8b1f7d6f653d : SUCCESS in 7m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6c9432f67a2a4a25b9e38436489ea23e : SUCCESS in 13m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9aae66ae81cb493ead888499046e6f77 : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c308daf256d4fc5aa79213737c7ccb2 : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a254d354fe6e46bbbc7ca5b9645369d2 : SUCCESS in 8m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8c402cac8d234335a1c89839926164e7 : SUCCESS in 10m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/11c1430d802b4a6195c8e5ac71703af3 : SUCCESS in 1h 49m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8be2466ea91646e181c90b7dc1070cca : SUCCESS in 1h 41m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ea8a17d4e9148b6a06125141f06f80a : SUCCESS in 1h 11m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/efac3c4806fe49a49015219342275377 : SUCCESS in 1h 36m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f8025ae736e249efb9ee606fe85bf9da : SUCCESS in 1h 27m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c1778e2d63a4ce4b584433bb82d698d : SUCCESS in 52m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c3d6876ab02481bb730be60a4213d77 : SUCCESS in 1h 35m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/520bbcb752684b3fa60c2d151f5e427b : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ead275dd352fbd7a70b5e53228462dff5b752a7b","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-01-30 08:10:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"98fa329633b9e7768f8427a8cb7988a0dd7684fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-01 06:07:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"963c4619d78f4f1c2801893b0e212dc1d8c2663e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-03-12 07:29:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b6ed6140e34d1d0910ee56d902a6d5e441bcd111","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-03-29 07:37:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1b1b2e3d0ebe1d6f72d2d811e2cc4b1ae5375434","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-01 13:43:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"423e2d0e609d0f7d6a01b6590e91b41f459e4d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 13:44:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c27e36fe02d0e8a3b7a23cf871b73e4fb6143689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:15:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b15595788c47228f112944499da688\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3fbd64542674e1198f630532a12ed36 : SUCCESS in 1h 17m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89124d08ab934f7c9b1ba99b91a2f66e : SUCCESS in 52m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c4efec339854895a4455f7ed2e7bb24 : SUCCESS in 1h 48m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8da888e39e2c47bcbcc2d2f6cecd0613 : SUCCESS in 1h 04m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe0b5dbd0b946b882f775db67b3befe : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5e83a2cc1ba4b2bb373fef07272fc44 : SUCCESS in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae0dbe1f1db241c7b0654ee9d6881fad : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001e56684570419b9bd7324ac7963fe3 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d47192c0b22498b8a569186d9a6c6c0 : SUCCESS in 3m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/591888cfb6b443b88224f97cfa4ca36c : SUCCESS in 9m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/56d4eb79a9ac4b18a8e49bad6b7f9094 : SUCCESS in 6m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/996c25a10fb748019f6c087c906df65a : SUCCESS in 2h 05m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab306dec19ee41beb9ab911c0968e758 : SUCCESS in 2h 25m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96d47cb897494b95aa7e733b5618906e : SUCCESS in 1h 13m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/727e528b8b0a42c2918c029cdb0981dc : SUCCESS in 1h 01m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3a4e589551c54b97bf37546d182c31c2 : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b45228dc35f78a0b745509b3df08e701a3a7a1ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:15:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"073df97c79af7f3c196a0f9459eb8ce60ac7805d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:17:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82263546a39f435993ef0275e4d80d67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4700830398d414cb300a9f5a1200afd : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a4c324d366e9cc262f0301ae9f5ce749c996507","revisions":{"9e1544ee442131f6c0e9c4e3a0acaccf9414709b":{"kind":"REWORK","_number":1,"created":"2019-09-12 15:46:08.000000000","uploader":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"ref":"refs/changes/87/681787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/681787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/681787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/681787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/681787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/681787/1"}}},"commit":{"parents":[{"commit":"73fefddd969e75e2f2d82866f4b8b9eeca488770","subject":"Add periodic job to prefetch images into cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73fefddd969e75e2f2d82866f4b8b9eeca488770"}]}],"author":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-09-12 15:42:20.000000000","tz":120},"committer":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-09-12 15:42:20.000000000","tz":120},"subject":"Make glance-manage honor config-dirs","message":"Make glance-manage honor config-dirs\n\nglance-manage is loading config from a couple of different config files\n(glance-registry.conf, -api.conf and -manage.conf). So it should be\nlooking into the corresponding directories as well.\n\nChange-Id: Ib47e96f23277cacbe7b8b9bc9907ea14aca69d75\nCloses-Bug: #1843772\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e1544ee442131f6c0e9c4e3a0acaccf9414709b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e1544ee442131f6c0e9c4e3a0acaccf9414709b"}]},"branch":"refs/heads/master"},"0a4c324d366e9cc262f0301ae9f5ce749c996507":{"kind":"REWORK","_number":2,"created":"2024-01-28 17:04:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/681787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/681787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/681787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/681787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/681787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/681787/2"}}},"commit":{"parents":[{"commit":"f429b53e0474553992f43c7045498156ee36d5cb","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f429b53e0474553992f43c7045498156ee36d5cb"}]}],"author":{"name":"Ralf Haferkamp","email":"rhafer@suse.de","date":"2019-09-12 15:42:20.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-28 17:04:04.000000000","tz":540},"subject":"Make glance-manage honor config-dirs","message":"Make glance-manage honor config-dirs\n\nglance-manage is loading config from a couple of different config files\n(glance-api.conf and -manage.conf). So it should be looking into\nthe corresponding directories as well.\n\nCloses-Bug: #1843772\nChange-Id: Ib47e96f23277cacbe7b8b9bc9907ea14aca69d75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a4c324d366e9cc262f0301ae9f5ce749c996507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a4c324d366e9cc262f0301ae9f5ce749c996507"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
