)]}'
{"id":"openstack%2Fglance~682652","triplet_id":"openstack%2Fglance~master~I71825cd5317d458fa30287b78cd1030cbc457dbd","project":"openstack/glance","branch":"master","topic":"rm-compute-trust-metadef","hashtags":[],"change_id":"I71825cd5317d458fa30287b78cd1030cbc457dbd","subject":"Remove OS::Compute::Trust metadef","status":"MERGED","created":"2019-09-17 13:46:59.000000000","updated":"2020-01-28 05:43:58.000000000","submitted":"2020-01-28 05:42:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682652-1580190154062-04bc3352","meta_rev_id":"815c2fda60f9ca18b677a706d594a38349ba7c26","_number":682652,"virtual_id_number":682652,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-17 15:33:20.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 05:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-02 11:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-01-28 05:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-01-28 05:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 15:33:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 11:07:30.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-01-28 05:23:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-01-28 05:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cf78d0455fb2de4e430588ac4db3ec833d05056","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-17 13:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e538996c8571813100d42c03173d1ff8f89912b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-17 15:33:20.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/52/682652/1/check-not-vote/ext-glance-zuul/c47f75b : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"09f17295df747b43aa6f63f64c939a69605c39c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 22:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f00abdab41674f26806653c827e2f5e0 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88f574dc631f40a2b54c6ccddd506abe : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b49d42c4894164869635c0925a9d81 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9f26b7e604c4f78959ecee2311a58a9 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d276ac398e468e93fe8b5c157716f7 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e57f099b6aa4af58487062a88d9eac7 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927f7bbf5c564d9fba4b84827fca5d8e : SUCCESS in 7m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c07fbfb5486b446ca20ae959066ec712 : SUCCESS in 8m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ffe87c48e40e44dfa3021e35cc088123 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"88811da66f372948ee3b782b2c6d185536b6619e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-02 11:07:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f04a330dcc6d4b4a3b0dfd816d5fdd2aa5fc0e47","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-28 05:23:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a67d8170297f16d66e8085f5a0e8dc94fba7e316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 05:23:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b310e2cda4f3883c79a8ab01d3b45d80ce591df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 05:42:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b9bd4b294414c5ea404b4f3a1f6143a : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2c801eb8ac74ef381d17643f3298b2d : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3eecfbf876d40208830108739f6417e : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18c4b10530344f81982ea050ba6f68a4 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4ef06e1efa14c43b44816da23e40a0e : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ea00966784452b96f222b0c982d421 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ce3bcdd5dab4d618023a5e49ea95853 : SUCCESS in 18m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e3091433971e4c5283ee33b292ad39a1 : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"98f8dbe7bc2df8098dd9a96979282b780f5d6c59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 05:42:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52188daa29ba2db7b5340f527d8385ec11f43e12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 05:43:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6f567bc9a1b1467bb9e8cf3440bd58f0 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a01d4471e324c659accae972553f8a6 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b78abdfa0ebc159263e9470a8636618927893a2b","revisions":{"b78abdfa0ebc159263e9470a8636618927893a2b":{"kind":"REWORK","_number":1,"created":"2019-09-17 13:46:59.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/52/682652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/682652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/682652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/682652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/682652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/682652/1"}}},"commit":{"parents":[{"commit":"d9546dce754c68084944ec01aaabd33790f8990a","subject":"Images API version bump and config sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9546dce754c68084944ec01aaabd33790f8990a"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 13:45:52.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-09-17 13:45:52.000000000","tz":-240},"subject":"Remove OS::Compute::Trust metadef","message":"Remove OS::Compute::Trust metadef\n\nThe nova TrustedFilter scheduler filter was removed in the Queens\nrelease [1] so the compute trust metadef is now just noise so we\nshould remove it.\n\n[1] https://review.opendev.org/506864/\n\nChange-Id: I71825cd5317d458fa30287b78cd1030cbc457dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b78abdfa0ebc159263e9470a8636618927893a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b78abdfa0ebc159263e9470a8636618927893a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
