)]}'
{"id":"openstack%2Fglance~683114","triplet_id":"openstack%2Fglance~master~I3c1b86e73a25935f2a2ffd43731e2db36fbffa97","project":"openstack/glance","branch":"master","topic":"train_reverts","hashtags":[],"change_id":"I3c1b86e73a25935f2a2ffd43731e2db36fbffa97","subject":"Revert \"Correct the deprecation messages of local dir config\"","status":"MERGED","created":"2019-09-19 11:41:02.000000000","updated":"2019-09-20 06:41:36.000000000","submitted":"2019-09-20 06:40:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683114-1568961610194-d2cb535d","meta_rev_id":"aa46e92e8dabcb802d504c2f977e8d1299547c43","_number":683114,"virtual_id_number":683114,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-09-19 14:48:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2019-09-20 06:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-09-19 12:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 03:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-20 03:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 12:22:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-09-19 14:48:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-20 03:57:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-09-20 06:40:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a33e941a6341d0eb7b3a9a45d3cdef1a6cc2aaa","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-09-19 11:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f6c71439c392e5ac905217a0710b7bd83c08da0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-19 12:22:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90cbc5d62d04ca7ce5518bd8139158268e125792","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-19 14:48:13.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/14/683114/1/check-not-vote/ext-glance-zuul/b738dba : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":1},{"id":"65855422938f757484748dacbe2dca9f199bfafd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 01:09:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2771d89e12544556b50aff78b92da0ef : SUCCESS in 1h 40m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ec6580aed00d482191914fb48004a7e2 : SUCCESS in 1h 08m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2ffd8f18c5694df2a8ffd8cfb564c2c6 : SUCCESS in 1h 09m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f97daee464c48d492181615af03bb15 : SUCCESS in 1h 36m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64c5f4bec5b4457c842cfe324cb462da : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cdc22db80994d7a95d350626e929b2e : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e38551c281e4ab8b8f61b841a90025a : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6d7cc956424f2fbb04a5c42a3cc1a0 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80778941406142bcb6bca8a53e00561c : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c68f8a1a2a4f68854c6dfa9e7ea82c : SUCCESS in 8m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bdf06a04b0b14a15b67bba83c187cbdd : SUCCESS in 10m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/487963150d04490baffeb1907af1086e : SUCCESS in 10m 40s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/73ced65a87284cfdac42273ac00e0e01 : FAILURE in 4m 20s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2ad8e0cb54534430b73f9dfcc2e1e0e0 : SUCCESS in 4m 45s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/675d710b844f4c5e9c7a3c257a57b7cb : SUCCESS in 1h 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d871bf3da0fa5d394e5db08c70b4ac251cbb1d7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-09-20 03:57:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89b1259cd61d8eb5f7667d958f1e33a498da6170","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 03:57:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d5d3ec56811ad490de8e59153ed9650e8cf9a93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:40:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/459d86291add434ba07ea95f01d90b58 : SUCCESS in 1h 46m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d2831e0bb6e54385b7f2840fc0e90c23 : SUCCESS in 1h 01m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e75306b9477a4083ba809ad22b476226 : SUCCESS in 59m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/491dd9aef19d492691886ac4faf2568b : SUCCESS in 1h 19m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68f87177bcee4f8fa50c6776fec1bd81 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fbf6cf724d48028a6bd02633bd0ad4 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f9e85d5718f43d994ed37ac7c9092d4 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f054c053ab6d40c38aa53e6cddbcb65d : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac38efe402ae4d2cb1f50f3d697e50b2 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a69b47af03c4ee68fd1e2d35fdec7de : SUCCESS in 6m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b7d1de61241046cfaeb311c67fac11c2 : SUCCESS in 9m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5277c599b55344e68dd8e2fb76ad10ac : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"70b5bcfdcc6cdf69315f565eb5d02b3c90c3b616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:40:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e6af59c6ec918bb107cce86127692e0ad27f9e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 06:41:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e33203a4d414a439338c47a770a41fb : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ae92bdd30d356ac84b24e4321138b2ebee7d7f5","revisions":{"4ae92bdd30d356ac84b24e4321138b2ebee7d7f5":{"kind":"REWORK","_number":1,"created":"2019-09-19 11:41:02.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/14/683114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/683114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/683114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/683114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/683114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/683114/1"}}},"commit":{"parents":[{"commit":"53fbfc8c635b778bc4731a352fc2fa3197328e3c","subject":"Merge \"Add Train milestone 3 releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53fbfc8c635b778bc4731a352fc2fa3197328e3c"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-19 11:32:51.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-09-19 11:38:56.000000000","tz":60},"subject":"Revert \"Correct the deprecation messages of local dir config\"","message":"Revert \"Correct the deprecation messages of local dir config\"\n\nthis is not needed as 4265e61bc84ce9bd085a95d8734647f4875af740\nis being reverted also.\n\nThis reverts commit b639ed080d5b74fcc3081a09bd64dd4703dea5a4.\n\nChange-Id: I3c1b86e73a25935f2a2ffd43731e2db36fbffa97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ae92bdd30d356ac84b24e4321138b2ebee7d7f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ae92bdd30d356ac84b24e4321138b2ebee7d7f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
