)]}'
{"id":"openstack%2Fglance~683838","triplet_id":"openstack%2Fglance~master~Ia67f9d74ae0d52111af4b0ddd2a034df7fbc5233","project":"openstack/glance","branch":"master","topic":"bug/1763761","attention_set":{},"removed_from_attention_set":{"19779":{"account":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"last_update":"2026-07-07 19:13:54.000000000","reason":"Change was abandoned"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-07-07 19:13:54.000000000","reason":"Change was abandoned"},"6873":{"account":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"last_update":"2026-07-07 19:13:54.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-06-26 17:58:52.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ia67f9d74ae0d52111af4b0ddd2a034df7fbc5233","subject":"Add numa specific image properties","status":"ABANDONED","created":"2019-09-23 02:58:26.000000000","updated":"2026-07-07 19:13:54.000000000","total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a2762fc1d8d65916f4d74a87b662e69488516cb0","_number":683838,"virtual_id_number":683838,"owner":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-24 02:56:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":-1,"date":"2023-06-26 09:11:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":-1,"date":"2026-07-07 19:13:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 03:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-23 14:18:43.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-23 14:18:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-24 06:55:53.000000000","updated_by":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"reviewer":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"state":"REVIEWER"},{"updated":"2019-09-24 12:09:51.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-02-03 07:07:53.000000000","updated_by":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"reviewer":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"state":"REMOVED"},{"updated":"2023-06-23 20:10:50.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2026-07-07 19:13:46.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"e93c32fcf2295eb10d03789ca5f3ffacb5c79090","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-09-23 02:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e586fe871acb236a80e5f8f3247f65c50d06116f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 03:06:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9ad23acd3a4e0d8a38b6c365fa1e73 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9322811c8ee8472abf7590cc904d5a2a : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"31a271a28c8294d750e3ba3573b70ab61277fa7d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-09-23 14:18:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0b9a8d333273f7869254b468feefbba3169131ea","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2019-09-24 06:55:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f96e5927d2f1b4ccdc26915c466aa5071eab50c4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-09-24 12:09:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0c8aca0c9ff5fcfb699fc8051c2560b8254b8a4b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-09-24 13:57:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01de42d7019ea9341a1c681355786509443c7d49","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2021-02-03 07:07:53.000000000","message":"Removed reviewer Lucian Petrut.","accounts_in_message":[],"_revision_number":1},{"id":"2eadb11af3ba798e9a8be8d344aff2234a9f3222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-23 20:09:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c576f0f1cd42e455067570bf3e851fc3b5c4468","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-23 20:10:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"508001142aab1771ba7433d2014a94c68f554895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 20:18:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/349ccb85c4a6415a9d61ed1c6b850618\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0137a1fd19964e19a3dbcffee5f6f4cc : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790043269c154f79af59b62aa2ee81e7 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b728836b7fcab73d8a82ad65199290e1060d4a85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-24 00:45:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8c8a2728b3740fd223bebcc52c1d7c14f52c58ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-24 00:59:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/878d6df834dc4627aab9062ea2b4729d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fb1f50012cb4fbabe9cdbadd20c0ae4 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/407a9d0797944f0f84ed3b15760d997d : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/032766d0d45d463e92f1e4a9ac15458f : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c23f180284f307b486ff04930aecd07a3618f47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-24 02:56:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc981d4506c745cba6963fd32378da63\n\n- grenade https://zuul.opendev.org/t/openstack/build/0dd60e27048b4a869eccf14d13237d63 : SUCCESS in 41m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48f785f35b3848e29d11b6a2e4aac4c2 : SUCCESS in 1h 16m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f2a4e37709447fb88c5ea27ba429c87 : SUCCESS in 54m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ccd5a6a2b04390b050997fc4c5a951 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49d913773d2349bf916be1ba3cfd704a : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d30f09875e644f0c830f320dd0b0b1c7 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deeeb9d0e575482ea7bfc19340ad20e6 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aae2de46d814b3ab4fb6ec726f2a7ff : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bf16ab6e35f746699e9df452b0d4de3a : SUCCESS in 13m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bb08552c4494c48b895b49b5fe0f7e1 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0a67d4f30634161920f4e0acc78f067 : SUCCESS in 8m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7068fc34ed444bfb85950f7b62455da1 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d61cf76d16104b4aa13a316c8faa6124 : TIMED_OUT in 2h 03m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cfcbfbac701c453b8d4aa251eafefe8e : SUCCESS in 1h 23m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dc7f72b82cd54914a4d88ffd72e10b17 : SUCCESS in 1h 15m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2317bbd73b6446329be05cd3c8dfb172 : SUCCESS in 1h 46m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b96fc03e1dd0454aa4020f00b53f2727 : FAILURE in 1h 08m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02ce032dfaa64b1e800eae6cc45c12d1 : SUCCESS in 1h 07m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0161ab04a29431a944792070688cb8b : FAILURE in 2h 04m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/016b55f6f678414bb0d3e0efd7a68cde : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":3},{"id":"15f88eaff9f23744b2eb07186fe3d4ae179cd6f9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-06-26 09:11:48.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"06b1ed445ae076aa8427e78dfb8665c6b274ab0a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-26 17:58:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b95a88b8823183462ad11d0f56d84ad5559fba6","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 19:13:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2762fc1d8d65916f4d74a87b662e69488516cb0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 19:13:54.000000000","message":"Abandoned\n\nSee comment","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5120a932298a9f7226bce91074c124d119627775","revisions":{"f42562ebddc559a6c265dc51c42ac601d8039192":{"kind":"REWORK","_number":1,"created":"2019-09-23 02:58:26.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/38/683838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/683838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/683838/1"}}},"commit":{"parents":[{"commit":"da29e129fa90862ef810b957a5bb0ef5095274d6","subject":"Revert \"Add reserved stores to the sample config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da29e129fa90862ef810b957a5bb0ef5095274d6"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-23 02:56:21.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-23 02:56:21.000000000","tz":480},"subject":"Add numa specific image properties","message":"Add numa specific image properties\n\nAdd ``hw_numa_nodes`` to useful image properties list.\n\nChange-Id: Ia67f9d74ae0d52111af4b0ddd2a034df7fbc5233\nCloses-Bug: #1763761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f42562ebddc559a6c265dc51c42ac601d8039192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f42562ebddc559a6c265dc51c42ac601d8039192"}]},"branch":"refs/heads/master"},"f8ca068448e382eef5d92291b0870a260c222af0":{"kind":"REWORK","_number":2,"created":"2023-06-23 20:09:40.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/38/683838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/683838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/683838/2"}}},"commit":{"parents":[{"commit":"d47803613d501b18f9fa4f6aa3c899bd16f9ddde","subject":"Handle case for multiple cirros images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d47803613d501b18f9fa4f6aa3c899bd16f9ddde"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2023-06-23 20:03:19.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-23 20:09:07.000000000","tz":120},"subject":"Add numa specific image properties","message":"Add numa specific image properties\n\nAdd hw_numa_{nodes,cpus,mem} to useful image properties list.\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia67f9d74ae0d52111af4b0ddd2a034df7fbc5233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8ca068448e382eef5d92291b0870a260c222af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8ca068448e382eef5d92291b0870a260c222af0"}]},"branch":"refs/heads/master"},"5120a932298a9f7226bce91074c124d119627775":{"kind":"REWORK","_number":3,"created":"2023-06-24 00:45:03.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/38/683838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/683838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/683838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/683838/3"}}},"commit":{"parents":[{"commit":"d47803613d501b18f9fa4f6aa3c899bd16f9ddde","subject":"Handle case for multiple cirros images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d47803613d501b18f9fa4f6aa3c899bd16f9ddde"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2023-06-23 20:03:19.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-24 00:43:46.000000000","tz":120},"subject":"Add numa specific image properties","message":"Add numa specific image properties\n\nAdd hw_numa_{nodes,cpus,mem} to useful image properties list.\n\nCo-Authored-By: Cyril Roelandt \u003ccyril@redhat.com\u003e\nChange-Id: Ia67f9d74ae0d52111af4b0ddd2a034df7fbc5233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5120a932298a9f7226bce91074c124d119627775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5120a932298a9f7226bce91074c124d119627775"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
