)]}'
{"id":"openstack%2Fglance~685674","triplet_id":"openstack%2Fglance~master~I99a274af222147fcb5e9969bbeb896312b30aeee","project":"openstack/glance","branch":"master","topic":"fix-hw_scsi_model","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-07-12 13:21:32.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-02-26 09:29:35.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-06-28 14:38:41.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"2394":{"account":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"last_update":"2022-09-15 19:05:56.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-02-26 09:29:35.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"hashtags":[],"change_id":"I99a274af222147fcb5e9969bbeb896312b30aeee","subject":"Improve accuracy and clarity of hw_scsi_model image property","status":"NEW","created":"2019-09-30 10:23:46.000000000","updated":"2024-02-26 09:42:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":19,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"8f3b57f693aeb21a6fea6d997b0c5920037a5ec5","_number":685674,"virtual_id_number":685674,"owner":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-17 01:57:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2023-07-12 02:18:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":-1,"date":"2024-02-26 09:42:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2023-07-12 13:21:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 10:24:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-09-30 10:24:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-30 10:42:29.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-09-30 11:49:42.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-09-30 11:49:56.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-09-30 15:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 15:34:34.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"},{"updated":"2022-09-15 19:05:56.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-06-28 14:38:41.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"f2c331965d850b9b9cdad57d7b6330b960031600","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-09-30 10:23:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e12df71b3b1aa03368efe37105ebf32bce423e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 10:32:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f633a340a324952bb0a40c94f89df31 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a2d94d0cf746a4bf7992a1a8d40c43 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ad641e72856d1fc01e947045997fbaaad7da0e4d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-09-30 10:42:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nits better the the original but it is assume things that are not strictly true.","accounts_in_message":[],"_revision_number":1},{"id":"b66640f4c2f61ed611a660907b6e81620c10f18d","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-09-30 11:46:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10538d80422211e7692b6ff57da0f2b22562fb06","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-09-30 15:27:50.000000000","message":"Patch Set 1:\n\nSee also https://review.opendev.org/#/c/684825/","accounts_in_message":[],"_revision_number":1},{"id":"94e236031a90215ecbbadaaae113857b381a8f04","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-09-30 15:30:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37179f439d8466ce4fa20f0dff7f416c05292291","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-09-30 15:31:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33ea1bccaf2c46e89af0c5792512f2b8fddec266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 15:53:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffaf4576924e4d218553ee19d5d11352 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95596779a4064e008f421f0a9294c889 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f221b103ba93e1ed513704e39245d1a9e60d3514","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2019-10-01 15:34:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc652825a8e41dbce99886a13ecea312340bc71f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-15 18:58:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bbff16acd0b25fc8183b2d909fe86efd704cb191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-15 19:05:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"903d989eaa062833cb03410be662466e60acfce5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-15 19:05:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"16f7f9840de900e5305774b1c62fccd6b9529028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 19:13:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11c3de8b87745e6af09357233114274\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e34dfdacc004dd08439c3e262e3915d : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad77beab8914e5590e3cad20c4f75fd : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":4},{"id":"acba5590a26e3f944cdb17513463a03268ff35e5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-16 01:49:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"63ec0e7b8e0b433b4c7ec66b7fda74464256ea9b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-09-16 09:59:20.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5162637729b7b7a477dc6f108435118175cc4c6e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-17 00:27:45.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c592c0fb8f30d3cba49b7f15174349c492a6338e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-17 00:28:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ac7881c75d07d1d1ff4d90350f7f5ca2d75e2318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-17 01:57:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca4419506b84026bc059d0ca8e16085\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1b015f67a61b4f4ba0596acfbfbe67b9 : SUCCESS in 5m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/28ea11d7ce0d48d1aa48bebd7512fd86 : SUCCESS in 1h 00m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3db70a3407944491b30d1d4e959e46da : SUCCESS in 1h 08m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b53506b87d64fa3a3ad4080a7749443 : SUCCESS in 1h 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e439d41151f34ba69fa1487a0850ad8e : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca769e84adc64d3e8a0b0da32f27d944 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ffde37fc64e4e62b2be3726088a4947 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d086f5260e84de585e5cdd422bc1c76 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2320c7b78540496ba8ab5157ef8b6e82 : SUCCESS in 10m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9da578757d6444849bb20f4f3a9a851e : SUCCESS in 8m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/39501c30169e475993bfca4b33256b96 : SUCCESS in 9m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b0a71b5c91dc473bb2d6b8c8bcec1788 : FAILURE in 49m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/221e2c95cb7e457d9d081c955ebd654e : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6585cf275ec8450dbae5e65b44698a80 : SUCCESS in 1h 06m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cec70fe195f84648aa96e7d0399e38ff : SUCCESS in 1h 28m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcf61bd042a44e28b31745b910e92c8d : SUCCESS in 43m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f14d2cc7fbc4067a6f189ea65708afc : SUCCESS in 59m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/147b035bb98e42d0b2a841d62e9d5e04 : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":5},{"id":"ccfc7c35e1866feafe4f0899135c3be80abcf3b3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-06-28 14:38:41.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a9d9374ff21e84bfd1f4a0b6e9afaf79fc338147","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-12 02:18:06.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7b09e96fc7b23503bba7cd4069eb0ef75014d4fb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-07-12 13:21:32.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dd8189627662d70d625081e8c33fe5edac38381d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-26 09:29:35.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8f3b57f693aeb21a6fea6d997b0c5920037a5ec5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-26 09:42:54.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aad1771864aa7e68edcf60076bb118ce46e4b055","revisions":{"238668a0b768794e3a157600a24282216c66c959":{"kind":"REWORK","_number":1,"created":"2019-09-30 10:23:46.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/74/685674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/685674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/685674/1"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 10:21:31.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 10:23:23.000000000","tz":60},"subject":"Improve accuracy and clarity of hw_scsi_model image property","message":"Improve accuracy and clarity of hw_scsi_model image property\n\nMake it list all valid values, and clarify that the default of\nvirtio-blk is behaviour which happens when hw_scsi_model is not\nspecified, rather than hw_scsi_model having a default value.\n\nChange-Id: I99a274af222147fcb5e9969bbeb896312b30aeee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/238668a0b768794e3a157600a24282216c66c959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/238668a0b768794e3a157600a24282216c66c959"}]},"branch":"refs/heads/master"},"bbc1e758aebd6c85a469b9fa36c60db036066d5e":{"kind":"REWORK","_number":2,"created":"2019-09-30 15:30:11.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/74/685674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/685674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/685674/2"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 10:21:31.000000000","tz":60},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 15:29:58.000000000","tz":60},"subject":"Improve accuracy and clarity of hw_scsi_model image property","message":"Improve accuracy and clarity of hw_scsi_model image property\n\nMake it list all valid values, and clarify that the default of\nvirtio-blk is behaviour which happens when hw_scsi_model is not\nspecified, rather than hw_scsi_model having a default value.\n\nChange-Id: I99a274af222147fcb5e9969bbeb896312b30aeee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbc1e758aebd6c85a469b9fa36c60db036066d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbc1e758aebd6c85a469b9fa36c60db036066d5e"}]},"branch":"refs/heads/master"},"1b1b382903ddf4f71f2166ef358ac34f3a2ccc57":{"kind":"REWORK","_number":3,"created":"2022-09-15 18:58:54.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/74/685674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/685674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/685674/3"}}},"commit":{"parents":[{"commit":"3217a9fb716673fc429d436697b358e14ef7d434","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3217a9fb716673fc429d436697b358e14ef7d434"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 10:21:31.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-09-15 18:56:03.000000000","tz":120},"subject":"Improve accuracy and clarity of hw_scsi_model image property","message":"Improve accuracy and clarity of hw_scsi_model image property\n\nMake it list all valid values, and clarify that the default of\nvirtio-blk is behaviour which happens when hw_scsi_model is not\nspecified, rather than hw_scsi_model having a default value.\n\nChange-Id: I99a274af222147fcb5e9969bbeb896312b30aeee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b1b382903ddf4f71f2166ef358ac34f3a2ccc57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b1b382903ddf4f71f2166ef358ac34f3a2ccc57"}]},"branch":"refs/heads/master"},"e2ddcda10f017d25a48706e4d50a148c048ad945":{"kind":"REWORK","_number":4,"created":"2022-09-15 19:05:09.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/74/685674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/685674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/685674/4"}}},"commit":{"parents":[{"commit":"3217a9fb716673fc429d436697b358e14ef7d434","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3217a9fb716673fc429d436697b358e14ef7d434"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 10:21:31.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-09-15 19:05:00.000000000","tz":120},"subject":"Improve accuracy and clarity of hw_scsi_model image property","message":"Improve accuracy and clarity of hw_scsi_model image property\n\nMake it list all valid values, and clarify that the default of\nvirtio-blk is behaviour which happens when hw_scsi_model is not\nspecified, rather than hw_scsi_model having a default value.\n\nChange-Id: I99a274af222147fcb5e9969bbeb896312b30aeee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2ddcda10f017d25a48706e4d50a148c048ad945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2ddcda10f017d25a48706e4d50a148c048ad945"}]},"branch":"refs/heads/master"},"aad1771864aa7e68edcf60076bb118ce46e4b055":{"kind":"REWORK","_number":5,"created":"2022-09-17 00:28:18.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/74/685674/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/685674/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/685674/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/685674/5"}}},"commit":{"parents":[{"commit":"3217a9fb716673fc429d436697b358e14ef7d434","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3217a9fb716673fc429d436697b358e14ef7d434"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-09-30 10:21:31.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-09-17 00:27:52.000000000","tz":120},"subject":"Improve accuracy and clarity of hw_scsi_model image property","message":"Improve accuracy and clarity of hw_scsi_model image property\n\nMake it list all valid values, and clarify that the default of\nvirtio-blk is behaviour which happens when hw_scsi_model is not\nspecified, rather than hw_scsi_model having a default value.\n\nChange-Id: I99a274af222147fcb5e9969bbeb896312b30aeee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aad1771864aa7e68edcf60076bb118ce46e4b055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aad1771864aa7e68edcf60076bb118ce46e4b055"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
