)]}'
{"id":"openstack%2Fglance~685924","triplet_id":"openstack%2Fglance~master~I37350fc46191074d42edc73c2451eae73261957d","project":"openstack/glance","branch":"master","topic":"fix_api-ref_version_index","hashtags":[],"change_id":"I37350fc46191074d42edc73c2451eae73261957d","subject":"Set API version 2.9 CURRENT","status":"MERGED","created":"2019-10-01 10:07:49.000000000","updated":"2019-10-01 23:37:54.000000000","submitted":"2019-10-01 23:36:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685924-1569972994546-b42f53c0","meta_rev_id":"d0c9a81e7eed6c0f80d48b278319ece6825944c6","_number":685924,"virtual_id_number":685924,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-01 23:22:36.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 23:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-10-01 19:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-10-01 19:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 14:20:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-10-01 19:40:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-10-01 23:22:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 23:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db02ae5c9e9a8b664fa54ea7f42460fcbe0062ed","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 10:07:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eef200241b83228e5f13686623ef2508542e16a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 10:38:03.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, all tests passed in locals!!","accounts_in_message":[],"_revision_number":1},{"id":"347caec52279e8bc3c1e6b84e2d775a7dc36638d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 10:48:43.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/24/685924/1/check-not-vote/ext-glance-zuul/ed1f3eb : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a1b34f916a9daed2dc8b0d4bd414f80b409ed428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:02:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5fecdd4b828a4685bc7f1f98a2d17685 : SUCCESS in 4m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f5bb05e9e6734be38eeef795f0a0eb1e : SUCCESS in 1h 50m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1d4e8f6dd7b44d16848f8906b50502f7 : FAILURE in 46m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e31b8ae69774d1996ea5e92a73fb0bb : SUCCESS in 59m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c5697c8e53e44fbabe44f1b9655676d : SUCCESS in 1h 36m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/915c05b21c354501acfb183ba940d7e2 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68416cb781a84e06b72ccfdd8031b5b9 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6aab5e811d174d38bf6283338fb92270 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e42196a3959c4674904db8757cf16403 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/049dcc997cb44aafbc0f693af4afcac6 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8975c60bedc14307a405697ca2b5b03d : SUCCESS in 7m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f556ade4d586478b8f950b72eab17fa2 : SUCCESS in 10m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/eb7071ec67924cfcb07d9b164a9b6402 : SUCCESS in 10m 40s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/2c5d2f0f5664409baa2f9f4fbaaf3c4d : FAILURE in 4m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/75e2bc5832bf4a539d1549627286a1eb : SUCCESS in 4m 25s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd9dd1f8d0f343cca64e3963853fb4a3 : SUCCESS in 1h 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2979ff6bc8d582f386b125217639b73167bf6ed1","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 12:02:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b8450e397a0779aaf2123eefa6a0a9573d42c9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:02:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0132636143394e5f91f7c38823ffaf7e : SUCCESS in 4m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/adc6e2ac8c054e5c839edec4f06e8b1a : SUCCESS in 1h 57m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97f2c77b29904f679d766311ce6a8b2b : SUCCESS in 59m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4a0da1881e9943398703fc45dfbb2870 : SUCCESS in 55m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cdac523a470466fb63ee29f9e79b910 : SUCCESS in 1h 31m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/168a412a678d4d8097f27544dcb32438 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b886ff74d84d58b2780af5edcfb66f : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0a2c8e936c3466bbd5648ab5bbab9b7 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a356027194bf434ba996a3e0f4bd098f : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27e5d705aac84983b8aa070a3cb742fe : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f49fcab6210d486aa2581e3f47657b73 : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e2684a6a52f4909ba65bcde1a4ef6df : SUCCESS in 8m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/71b329ab7d5b44678dffa770d8165b4f : SUCCESS in 8m 33s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/9d2beafe309a448684fa7e6ee0c2c945 : FAILURE in 5m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/395f3ec48b1742edb9808ccf1490a8ea : SUCCESS in 5m 41s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47ff87b285c4491a8c4c3feb68e36cf1 : SUCCESS in 1h 06m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74497f5533d6f2248ced06bbc276f8b4cebafff2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 14:08:16.000000000","message":"Patch Set 1: Code-Review-1\n\nNeed to update api-ref/source/versions/index.rst\n\nadd:\n\n  **Train changes**\n  - version 2.9 is CURRENT\n  - version 2.8 is SUPPORTED","accounts_in_message":[],"_revision_number":1},{"id":"8923b352aa674587d483421edd86507307954780","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 14:19:02.000000000","message":"Patch Set 1:\n\nAlso, need \u0027- version 2.7 is SUPPORTED\u0027","accounts_in_message":[],"_revision_number":1},{"id":"5b192a82f312fe24fe9e5089c7a20ba1c7334348","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-01 14:20:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThe version history update will be done in a follow-up patch: https://review.opendev.org/685977","accounts_in_message":[],"_revision_number":1},{"id":"51a00458c60262754e5dabf4b66fa4102a2116ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 14:21:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8605186d6bd347345d4b740d3d90ad9101fceb9c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 15:57:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83cddb93f06e7763dfb60f096e22ed215b134d36","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-10-01 17:45:32.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9053ce06b35ca3835f34f459bec81bed1156d9e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:34:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d68acc86d354d92a2f2668ede1d9ccb : SUCCESS in 4m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4fc7c844c06c4fe8a54c3cf44f754954 : SUCCESS in 1h 41m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6fd6852ad9f84293a6e30b6f48360e3c : SUCCESS in 57m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f722bfabacaf42aeb1981122b85b9b2b : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a17f971cf9340b885396bd65f498259 : SUCCESS in 1h 24m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36dbfbb53ff14380bac0b89ac0ad9e7a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9097c99ca3448ae9466e5cf827cd107 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcbfe7639a70423b94b70e381824396e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/decc9377b67f440384f9ee167df4d7de : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aabe93da419546069565eae7c87832fe : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5378df7706234d8dbbeb4de2b69600a9 : SUCCESS in 6m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b0513709ce004a66becf41e7381628bb : SUCCESS in 8m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7df13f4835b44b1ab2473aa0f6d7b3a4 : SUCCESS in 9m 06s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/8b4ad6668e154950aee0c43f7842944f : FAILURE in 4m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2c7b793267aa4c76a759e16fafc9ea46 : SUCCESS in 4m 04s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bad96335b20a41f88c51ad59552ac12d : SUCCESS in 1h 07m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b170cbc19ec21afc86c9e261da9883b22dbb23f4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-10-01 19:40:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLets gate it!!","accounts_in_message":[],"_revision_number":3},{"id":"e3b8546016a475f924e3507ca773d858d6d7531a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:40:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70868a870f170b667ebf127eeaf15d592d5db8a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 20:42:17.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a88a871765b04f562f18d85272865e792994815b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-01 23:22:36.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/24/685924/3/check-not-vote/ext-glance-zuul/28be46f : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f9bbe1dbc31c1cc36127e4cbad6d51d624c06396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 23:36:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/201b397744f94a8b886784e4f39c5662 : SUCCESS in 6m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2f7147bbb3f1462a987e1639661d964c : SUCCESS in 1h 46m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/22142a232f7444cfb276bf1c7a7b6d37 : SUCCESS in 1h 15m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01885d49a9574ea4a1d6229c383d0b10 : SUCCESS in 59m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02ec5a595c604d8ea65f4680fa17469a : SUCCESS in 1h 29m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9f9aa2fc2674d13a05272bbc7f76df4 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed969d034bfc42179717a1e4204442ad : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b45090fac47b40fb9ec960aa4afe07bc : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76aa2e6ea18d44b1a7ef464b75cd33e5 : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12bb8a19098a475e8b7765064ede2d02 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60469557f1e4e41a1ae38bfb9b9cc08 : SUCCESS in 8m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d521acb2724b49679bae6dc53ab848dc : SUCCESS in 11m 31s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/146b9f47a2174f06a453713f36f6560d : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":3},{"id":"488245118f453dbd369d0f338912481d5372e4a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 23:36:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"39a04b5218e0c39db1f2d804cc8ed89e7646b6c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 23:37:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6ca9c8466aa94f0098c49ba81c9630d9 : SUCCESS in 51s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1068062fec04b358ec29bf8a4de86e9 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"515412b59f5b3af07a1787b9f8e85a4d656d3e1c","revisions":{"0b77c828c247890db9b4fa01da2a603c95de4e5c":{"kind":"REWORK","_number":1,"created":"2019-10-01 10:07:49.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/24/685924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/685924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/685924/1"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-10-01 10:04:15.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-10-01 10:04:15.000000000","tz":60},"subject":"Set API versions 2.9 CURRENT","message":"Set API versions 2.9 CURRENT\n\nThis change removes experimental status from Api versions\n\u003e2.7 setting 2.8 (multi-store support) to SUPPORTED and\n2.9 (caching) to CURRENT.\n\nNOTE: 2.8 will show up in list only if the glance-api service\nis configured with multi-store. The API versions list can be\nused to determine if the multi-store functionality is availabe\nin the endpoint.\n\nChange-Id: I37350fc46191074d42edc73c2451eae73261957d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b77c828c247890db9b4fa01da2a603c95de4e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b77c828c247890db9b4fa01da2a603c95de4e5c"}]},"branch":"refs/heads/master"},"98c3962ab8fe6880ccc5f29249eb754c546ae937":{"kind":"REWORK","_number":2,"created":"2019-10-01 15:57:02.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/24/685924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/685924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/685924/2"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-10-01 15:56:20.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-10-01 15:56:20.000000000","tz":60},"subject":"Set API versions 2.9 CURRENT","message":"Set API versions 2.9 CURRENT\n\nThis change removes experimental status from Api versions\n\u003e2.7 setting 2.8 (multi-store support) to SUPPORTED and\n2.9 (caching) to CURRENT.\n\nNOTE: 2.8 will show up in list only if the glance-api service\nis configured with multi-store. The API versions list can be\nused to determine if the multi-store functionality is availabe\nin the endpoint.\n\nChange-Id: I37350fc46191074d42edc73c2451eae73261957d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98c3962ab8fe6880ccc5f29249eb754c546ae937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98c3962ab8fe6880ccc5f29249eb754c546ae937"}]},"branch":"refs/heads/master"},"515412b59f5b3af07a1787b9f8e85a4d656d3e1c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-01 17:45:32.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/24/685924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/685924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/685924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/685924/3"}}},"commit":{"parents":[{"commit":"feb8ebd75bfcf1296dfa40e000a4a16eb2082971","subject":"Merge \"Add support for oslo.reports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/feb8ebd75bfcf1296dfa40e000a4a16eb2082971"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-10-01 15:56:20.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2019-10-01 17:45:26.000000000","tz":0},"subject":"Set API version 2.9 CURRENT","message":"Set API version 2.9 CURRENT\n\nThis change removes experimental status from Api versions\n\u003e2.7 setting 2.8 (multi-store support) to SUPPORTED and\n2.9 (caching) to CURRENT.\n\nNOTE: 2.8 will show up in list only if the glance-api service\nis configured with multi-store. The API versions list can be\nused to determine if the multi-store functionality is availabe\nin the endpoint.\n\nChange-Id: I37350fc46191074d42edc73c2451eae73261957d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/515412b59f5b3af07a1787b9f8e85a4d656d3e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/515412b59f5b3af07a1787b9f8e85a4d656d3e1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
