)]}'
{"id":"openstack%2Fglance~686461","triplet_id":"openstack%2Fglance~master~I3036aeeb40a95c9270578dc6d376fe5246981d69","project":"openstack/glance","branch":"master","topic":"bug/1846548","hashtags":[],"change_id":"I3036aeeb40a95c9270578dc6d376fe5246981d69","subject":"Allow DB schema creation on MySQL 8","status":"ABANDONED","created":"2019-10-03 18:33:12.000000000","updated":"2019-10-04 15:16:12.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3642d9893ea4fa903c0f5b080c06e18dcbb204e1","_number":686461,"virtual_id_number":686461,"owner":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-03 20:16:42.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-03 20:14:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-04 15:15:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"value":-1,"date":"2019-10-04 09:56:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-04 15:15:55.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 20:14:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 20:16:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-04 09:56:03.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"}],"messages":[{"id":"4c7331e1758566ecbbd1ec356580342692ed5f52","author":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"date":"2019-10-03 18:33:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48d3e4452cb1a8c2f0648a9b653cc59290355935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 20:14:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/86405681f3be4fdfb94bb9879361d90b : FAILURE in 1h 29m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/47d5b102b43447808f5d335978f85ef6 : SUCCESS in 1h 03m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e8dc3212f31e46839430a26e2e0dbd33 : SUCCESS in 1h 06m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96b599c29faa4cfe9909025bb3d87c3c : FAILURE in 1h 22m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/065dabcf34fa48adaa72cadfee6e3274 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05fad0da919542ee949c144666e0fdae : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a2c870827f84a32b990fa93b86e5c3e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4405780564e74ca48a49558167325fbc : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cbb3f2f8c4b45cf854834ff5ba575dc : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db9cd1e20ae433b960cb6bd9562f6dd : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b66e514245024a9899bbc78d82ea87f3 : FAILURE in 9m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/943eeca827e747d5bef6296ca36d6973 : FAILURE in 14m 15s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/06bccaae912b4392b1ac582554612e37 : FAILURE in 4m 35s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7d8643c4355441ba93ba166d076c20b0 : SUCCESS in 3m 57s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db6d98f5f8d84c90b04658fb73ba8438 : FAILURE in 1h 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea4a00bf5f09d1a794fededa89b88f816fde9979","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-03 20:16:42.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/61/686461/1/check-not-vote/ext-glance-zuul/cc4022e : FAILURE in 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0b41d3c857c9a1d34552be1166268fe2e893178f","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-10-04 06:18:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8202b892351d9f99b3cf812c2cbfeeb84647cbc0","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-10-04 06:22:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"437657bda6548e8905c9fd4dacd00494c1e1740b","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-10-04 09:56:03.000000000","message":"Patch Set 1: Code-Review-1\n\nSQLAlchemy has fixes for this issue in 1.3.x series; as a result I don\u0027t think this change is required for glance.","accounts_in_message":[],"_revision_number":1},{"id":"b6a130469f9eac3ee7c31c6c691b79b67ead40e8","author":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"date":"2019-10-04 15:15:55.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nI agree with James Page. Abandoning this change.","accounts_in_message":[],"_revision_number":1},{"id":"87d40a7bad2f273e0e449ab8a91d8cac2bae45e0","author":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"date":"2019-10-04 15:16:12.000000000","message":"Abandoned\n\nDifferent solution required.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07ac4b6b6646ddbec2b0531da8da3898fb608eb1","revisions":{"07ac4b6b6646ddbec2b0531da8da3898fb608eb1":{"kind":"REWORK","_number":1,"created":"2019-10-03 18:33:12.000000000","uploader":{"_account_id":20805,"name":"David Ames","email":"david@davam.net","username":"thedac"},"ref":"refs/changes/61/686461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/686461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/686461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/686461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/686461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/686461/1"}}},"commit":{"parents":[{"commit":"99938b8afeebcdafc7baa07b73b7762abcc646a7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99938b8afeebcdafc7baa07b73b7762abcc646a7"}]}],"author":{"name":"David Ames","email":"david.ames@canonical.com","date":"2019-10-03 18:29:11.000000000","tz":-420},"committer":{"name":"David Ames","email":"david.ames@canonical.com","date":"2019-10-03 18:32:37.000000000","tz":-420},"subject":"Allow DB schema creation on MySQL 8","message":"Allow DB schema creation on MySQL 8\n\nGlance currently fails db_sync when using MySQL 8. The root cause issue\nis that \"member\" is a reserved word in MySQL 8 [0]. Glance is attempting\nto create a table named \"member\" but is not adequately quoting the table\nname. Reserved words should be back tick quoted: `member`\n\n[0] https://dev.mysql.com/doc/refman/8.0/en/keywords.html\n\nChange-Id: I3036aeeb40a95c9270578dc6d376fe5246981d69\nCloses-Bug: #1846548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07ac4b6b6646ddbec2b0531da8da3898fb608eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07ac4b6b6646ddbec2b0531da8da3898fb608eb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
