)]}'
{"id":"openstack%2Fglance~689018","triplet_id":"openstack%2Fglance~master~Ia76b107a137a39716253631e5a8d294f7ee9614e","project":"openstack/glance","branch":"master","topic":"opensuse-150","hashtags":[],"change_id":"Ia76b107a137a39716253631e5a8d294f7ee9614e","subject":"Switch to opensuse-15 nodeset","status":"MERGED","created":"2019-10-16 19:41:24.000000000","updated":"2019-10-29 22:15:26.000000000","submitted":"2019-10-29 22:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689018-1572386814733-e7a1f455","meta_rev_id":"3bbbe5dd2e482713424120848292554ed7e80d80","_number":689018,"virtual_id_number":689018,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 22:06:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-29 16:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-29 16:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-29 16:59:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 16:52:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-29 16:59:44.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-10-29 22:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a619450717beb8c62ccb9730eefb287679c72855","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-16 19:41:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"845d410556fb3970b88252868a050d0100a8d1e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 21:55:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bc608fbe82c40838ca8e3fc915de44f : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24dc929dc8ea4c7db91b5f47f64380b9 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3aa82d467a364cfdbef2eb221f98d0cc : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55a26aed3b624331a7b51a629cbf7e8b : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7f67315e0c0417d8cc5d2a93f64c133 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec558e42fa9d49ffa9d72496ff67a646 : SUCCESS in 5m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/71755c902b934877a499d07781657c01 : SUCCESS in 11m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8393c93fc09d48d180a0559783b03883 : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"069b008d20ebc12f500b0942b0b8e00366e2a602","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-29 16:52:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a534ede4a8af1d9d6b063ee0de6f8af3b614a38f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-29 16:59:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMakes sense to me.  Thanks for the detailed commit message.","accounts_in_message":[],"_revision_number":1},{"id":"bf190174c3e23bf95d346bd8cc0b0ad777842d63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 17:00:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62c9cafb8059538832dd5089adbc8edbacebd183","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:06:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e1c212663357daf42f0cb6feb361d101cbf3811","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:06:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d6ec1941efe4149a82c1a0fc958924b : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2da147222b495e90647485a774e4db : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88dfba2e8ad44cf88e839c973b4c5a45 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ddaf27d9f644b78a3c75819ab78470 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/251a73eb0e8a4f2880e31be7490196b0 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e865e4739e47c4b3a199658f0020ee : SUCCESS in 6m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/37c050d7e90f4c268e3f8ab6aa39b738 : SUCCESS in 9m 43s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a2f8a795062347cf8b2c7eda8c2f6280 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fb42d776515cb71527370862d03f017d3cae6703","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:15:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e3d8df76614e01b43e1b8389cc7a4a : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e45d5e3b9259620914ff3e3c5fa67dd23de1879c","revisions":{"e45d5e3b9259620914ff3e3c5fa67dd23de1879c":{"kind":"REWORK","_number":1,"created":"2019-10-16 19:41:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/18/689018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/689018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/689018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/689018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/689018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/689018/1"}}},"commit":{"parents":[{"commit":"99938b8afeebcdafc7baa07b73b7762abcc646a7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99938b8afeebcdafc7baa07b73b7762abcc646a7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:41:09.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:41:09.000000000","tz":120},"subject":"Switch to opensuse-15 nodeset","message":"Switch to opensuse-15 nodeset\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nThe new job tempest-full-py3-opensuse15 use the opensuse-15 nodeset,\nchange tempest-full-py3-opensuse150 to tempest-full-py3-opensuse15.\n\nChange-Id: Ia76b107a137a39716253631e5a8d294f7ee9614e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e45d5e3b9259620914ff3e3c5fa67dd23de1879c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e45d5e3b9259620914ff3e3c5fa67dd23de1879c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
