)]}'
{"id":"openstack%2Fglance~691308","triplet_id":"openstack%2Fglance~stable%2Fstein~I8a763eff7c6524969b1fabf98ab0a5624954d8b3","project":"openstack/glance","branch":"stable/stein","hashtags":[],"change_id":"I8a763eff7c6524969b1fabf98ab0a5624954d8b3","subject":"Periodic jobs are failing for python 3.5","status":"MERGED","created":"2019-10-25 13:33:32.000000000","updated":"2019-11-22 08:03:15.000000000","submitted":"2019-11-22 08:00:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691308-1574409626202-c4d0231e","meta_rev_id":"a476626f74494938212ec49f9142a171fdc42a68","_number":691308,"virtual_id_number":691308,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 08:00:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-11-20 11:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-11-20 11:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-11-20 11:28:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-11-22 05:05:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2019-11-20 11:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 19:23:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-11-14 14:47:28.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-11-22 05:05:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-11-22 08:00:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2580a9462101b40f2e628254d8e8119ef746035d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-10-25 13:33:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1964f297028c5ea0d7561d33a4a9ade416645ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 15:05:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/df4a4a6e03be4a0daa04e6a04816791a : SUCCESS in 1h 06m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcd6136a6d8741e2ba5b39f914310a6b : SUCCESS in 1h 29m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8bff921cc1242ec937c9ab53355b4a3 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc79a8ed2fe406387e038db0acee615 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c11fdfab19cd436a90af9152fa04e321 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31b3e697dd8d42ef9106cb6b5183699b : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c185b4f19e48414baf7bce36017095a4 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f28eaf04825a4701906077436eec1c3b : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df477a0766942a3828b7f43d7dbea7c : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ae7f02e08c4c4f9832c535b6c7d92f : SUCCESS in 9m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/24e432a5ddc84587903cbe9a51693734 : SUCCESS in 9m 33s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6e2856e03e60490ca14f0c20ed98c9d9 : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"eed8b02e6d0e6898324a4ee5f326c98df9c20ac9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-06 02:31:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8916a4e184e28e36c0b7f85603e1cd7f34b95331","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-12 19:23:11.000000000","message":"Patch Set 1: Code-Review+2\n\nStein is supposed to be tested against a minimum of python 3.6 [0], so this backport appears correct to me.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/stein.html#python-runtime-for-stein","accounts_in_message":[],"_revision_number":1},{"id":"8c4a13a112d68c6f1326bc230724098762653cba","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-11-14 14:47:28.000000000","message":"Patch Set 1: Code-Review-1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53a6e850e4336f2c1b46fcc44139d64aa68d2966","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-11-14 14:55:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nActually the py35 envs were left behind already on the patch to stable/train we need to address too. Lets keep it separate for clarity","accounts_in_message":[],"_revision_number":1},{"id":"acbc2462548b9dcea1aa5af78109692de144f6eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:56:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3d19f74928e022e4f3e3be212a4ac5de44a1d18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:16:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/87de15e54a504d38bf778bf307ec7552 : SUCCESS in 1h 03m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e50341edbe534b189054830914dcdde6 : FAILURE in 1h 17m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a95f7d747a3d415ca985af43df0fc850 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70c43d04a0f491dbe4f83469cfdc05c : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1c0d853cb484c0bbd50c1aa49f2658a : SUCCESS in 7m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/03f0008562bd46d6b9a7d61e2da3e8ff : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fe317a395024922bd13714417673680 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02ae3a221f9d49589b93f57f8a1c1a25 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36dc6a18297646b9a150f52048fba056 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab9e60c778f489a990fce09a585dbb2 : SUCCESS in 10m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/67e9d32cc16846ffaf059e1b33f19d03 : FAILURE in 9m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7d82dcb5c28b4fd7ba98cbf4716aacd7 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7d16cdb3e921dcdf52931dca1f6aee6cd0217173","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-14 16:50:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be5a1cbf6cb52ebf688f9c724425c644feba9053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 18:42:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f22452f3a7c049e3b101a9a63612f4d7 : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ed633f881de45af953df039a6ad223a : SUCCESS in 1h 34m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae82e58e5f224e478b15be37895f6e7b : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59dae83859414e48b5d26c97619e9633 : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5889e43700454520b5f11cc331d45892 : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e99df14109514343b230d3566e6c91b5 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f430a811f8947708408b522cf09cee9 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ee8a940ca9443879d68d1e431ccb4a2 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8bd97f9a72476b80b23b0e1a372789 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7894e2a7556a414ea0275bb7151e58da : SUCCESS in 11m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dafb8acc5d6c4f5c8dd3710b32bae37f : FAILURE in 10m 50s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/1d284a4a1f9e48e786127ce49b0c6528 : TIMED_OUT in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0321706d79ff2b9ef1b6a4b34d128ea6332a1b6d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-14 19:06:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00cce3f9b56b1f05af00076691af7bb089347260","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-14 19:07:53.000000000","message":"Patch Set 1:\n\nThis subunit parser error now consistently causing failure in stable stein branch.","accounts_in_message":[],"_revision_number":1},{"id":"8a076beb6cb44e04dd763c842921fddf08111f30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:08:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2d9814b8bd374591b1bc1f66d7d6d0e1 : SUCCESS in 54m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3d22ba56d4745238135bbebbdcbdef7 : SUCCESS in 1h 22m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8974283c3cff463ca53c06f1ef8bf78d : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2610a1c0ebb745be841c57cc237bc5b3 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/060218628df24450a62b36127787ffcd : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a40ac775a1d149d792f419af1bbd1d26 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af9086b901a141ca81e7545436e7ec3e : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b7922c097ff461fbfe75bf5f4bc7379 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff376fbf7914f1c99ba61af2a389ee6 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5216b5002dd3470b9c3f48d909df7b61 : SUCCESS in 7m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4b9bb86a2613435c89ff04073854205f : FAILURE in 9m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d8b61f8f4a434f4e952f3a73556a9162 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b7ae46e7da5670507df96f2e248a56fcb08a3b74","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-15 05:20:22.000000000","message":"Patch Set 1:\n\nIf I comment out https://github.com/openstack/glance/blob/stable/stein/glance/tests/functional/__init__.py#L930 and https://github.com/openstack/glance/blob/stable/stein/glance/tests/functional/__init__.py#L1298 then subunit.parser error doesn\u0027t occur at all.","accounts_in_message":[],"_revision_number":1},{"id":"e7fa31fb9dff0f9b435abc7dc37c3f6cb3a7c454","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-15 08:53:17.000000000","message":"Patch Set 1:\n\nThe questions now are: why did it work ~20 days ago? Are other branches affected?","accounts_in_message":[],"_revision_number":1},{"id":"ef1912bd2ba0a0ea2c0ba003320257bc0d78d80e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-15 09:01:01.000000000","message":"Patch Set 1:\n\n\u003e The questions now are: why did it work ~20 days ago? Are other\n \u003e branches affected?\n\nNo Idea, It is not failing on master branch and failing consistently on stable/stein branch. So far we are not able to figure out the reason behind this error, but due to length of log size this error occurs. Not sure why it was working 20 days ago and not now.","accounts_in_message":[],"_revision_number":1},{"id":"9dfa60fd9ac6e3dbfb90869e5072dfbd7e0d5a3b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-15 10:32:40.000000000","message":"Patch Set 1:\n\nThe difference was that functional was using python2, and now it\u0027s using python3 (3.6). But the base functional for this version should still be using python 2.7, afaik. The switch happened later.","accounts_in_message":[],"_revision_number":1},{"id":"f16320ed04b30d347915e404c91977003cd82314","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-15 10:37:25.000000000","message":"Patch Set 1:\n\n\u003e The difference was that functional was using python2, and now it\u0027s\n \u003e using python3 (3.6). But the base functional for this version\n \u003e should still be using python 2.7, afaik. The switch happened later.\n\nWas this same 20 days ago when it was passing?","accounts_in_message":[],"_revision_number":1},{"id":"51393cd7cb8d4bc014f16f36efaf212a43086278","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-15 10:43:30.000000000","message":"Patch Set 1:\n\n\u003e \u003e The difference was that functional was using python2, and now\n \u003e it\u0027s\n \u003e \u003e using python3 (3.6). But the base functional for this version\n \u003e \u003e should still be using python 2.7, afaik. The switch happened\n \u003e later.\n \u003e \n \u003e Was this same 20 days ago when it was passing?\n\n20 days ago they were using python 2.7. Hence the difference. See: \nhttps://zuul.opendev.org/t/openstack/build/24e432a5ddc84587903cbe9a51693734/log/job-output.txt#510\n\nI\u0027m still don\u0027t know what lead to this change.","accounts_in_message":[],"_revision_number":1},{"id":"319b47913a7e4069d260627da3aa06ff85336a77","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-18 06:02:27.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e The difference was that functional was using python2, and now\n \u003e \u003e it\u0027s\n \u003e \u003e \u003e using python3 (3.6). But the base functional for this version\n \u003e \u003e \u003e should still be using python 2.7, afaik. The switch happened\n \u003e \u003e later.\n \u003e \u003e\n \u003e \u003e Was this same 20 days ago when it was passing?\n \u003e \n \u003e 20 days ago they were using python 2.7. Hence the difference. See:\n \u003e https://zuul.opendev.org/t/openstack/build/24e432a5ddc84587903cbe9a51693734/log/job-output.txt#510\n \u003e \n \u003e I\u0027m still don\u0027t know what lead to this change.\n\nShould we contact infra to understand this change?","accounts_in_message":[],"_revision_number":1},{"id":"a08b234754d443705d3d4beb16c45e1430937dec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-18 06:02:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b960d85e9050379ccafe85655e2ec4409301b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 08:49:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac63a76182194235893848d6cd45805d : SUCCESS in 45m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f29184b45cfe4caeb7f5dc239e9fab7a : SUCCESS in 1h 23m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1730e13437e34396aa33e2c297bfdfe5 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6249563118f747cca64c6b0f425d1a74 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1cf9529d4e984c8a9bd3e65dbef67f55 : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8298fecf23394536ac24bffbcbdda997 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e6281485d1e4e6da579abc6fe94b350 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f2b53b5bde04c5a8f640ba1eac0f1f6 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd16dd277034a1fa2c06f75924f7f1f : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec989f1e6e947f2bce2dc8a3328d1b2 : SUCCESS in 9m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/52e3cbfd54904bf9ac0993eaa4fcbf55 : FAILURE in 9m 30s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/d68105183a3745abb252acd973342fbf : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2361ecf73530b6ed4a72ac52917a3552eb9a4a3c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-18 08:53:03.000000000","message":"Patch Set 1:\n\n\u003e \n \u003e Should we contact infra to understand this change?\n\nI did, and there is a change which I think it\u0027s a regression, but not all infra people agreed on that. It will be discussed during the next infra meeting tomorrow:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-infra/%23openstack-infra.2019-11-15.log.html#t2019-11-15T11:10:57\n\nhttps://wiki.openstack.org/wiki/Meetings/InfraTeamMeeting#Agenda_for_next_meeting","accounts_in_message":[],"_revision_number":1},{"id":"b78912861c7382284610a04cb74e08e2bfcc2903","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-18 08:57:30.000000000","message":"Patch Set 1:\n\n\u003e \u003e\n \u003e \u003e Should we contact infra to understand this change?\n \u003e \n \u003e I did, and there is a change which I think it\u0027s a regression, but\n \u003e not all infra people agreed on that. It will be discussed during\n \u003e the next infra meeting tomorrow:\n \u003e \n \u003e http://eavesdrop.openstack.org/irclogs/%23openstack-infra/%23openstack-infra.2019-11-15.log.html#t2019-11-15T11:10:57\n \u003e \n \u003e https://wiki.openstack.org/wiki/Meetings/InfraTeamMeeting#Agenda_for_next_meeting\n\nCool, thank you for the update.","accounts_in_message":[],"_revision_number":1},{"id":"bd8948bd09e6855944eae96b6ac281c4fd846b6c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-20 11:28:29.000000000","message":"Patch Set 1:\n\nUpdates: you can read the logs here:\nhttp://eavesdrop.openstack.org/meetings/infra/2019/infra.2019-11-19-19.05.log.html#l-102\n\nThis e-mail summarizes the issue:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-November/010957.html\n\nAs the fix was already applied to stable/train (then master) as https://review.opendev.org/#/c/665153/ , I\u0027m going to backport that patch as well at the bottom of the stack of patches. I\u0027m also going to add the final patch which drops py35, as requested (another backport, https://review.opendev.org/#/c/664475/ ).","accounts_in_message":[],"_revision_number":1},{"id":"0a0218fa40559c77e532656f0270a82265611a64","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-20 11:28:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"97c97474b38de441557469164adc8b558b6e678d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:12:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/85a4e5956a814d75975728c1de13eb4d : SUCCESS in 51m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c73b8078b6242f9bbd2879fd3ba6888 : SUCCESS in 1h 07m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bee4b0d506d45d9981fed3c62d82cc7 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54fcfb470e854b8a910d08e297efd6bd : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5480a4eb7c244d23871cc032f20ab741 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3a4e6c8b99724fa7942cb5ce5b468ebf : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698cf4771c0749698b29b8c6f07aba8e : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e0665d26e084b18af01e191d2e2e675 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a812448ddcbe457280aa6f06efb464cd : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/355ac96ab5f143beac0880dd2360f811 : SUCCESS in 9m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/39ad643962e04cce83c078b6df4c96c7 : SUCCESS in 7m 45s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7da815f9bb0c41e98a29657b0116931f : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":2},{"id":"572a085fa6b374ac36526818d495132375ecc2db","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-22 05:05:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94552c4a120e763d5c07be66cf6cc6dda96c9c9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 05:06:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fca9c6cda38ce0b00f92a0df14491265f1eae309","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 08:00:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2764aaa781fd43cca38068ea3c38420e : SUCCESS in 56m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/891d43513f434ed3b99136c4f2d2a142 : SUCCESS in 1h 06m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeb21834e9994fff9f292dae6ba15f66 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd04f1f7f2b45389d0b34ad1fc1dcae : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5d658b7c70f49729553c8d240ba6f66 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/933b0290eb4249a996e897e27b8560aa : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5e4dbef12f248b58ecd16d80e09fe16 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77708d7eaf4b4718939219858fe3050e : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9400d2f7c3dd40baa5e6211dcbf1c770 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ff60e733eb45f8afc9517573e7bd6f : SUCCESS in 9m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/76ae5cde2b354ef2854ae1e49135be5a : SUCCESS in 10m 01s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/df1359b735ac47c28d95012027192f0d : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1773ca5415dbae9416eb9db24975f7183ff99d37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 08:00:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a476626f74494938212ec49f9142a171fdc42a68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 08:03:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b6b37c4e5f45bebe47fa225a43e6b1 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1808595bee1d46ad8182d7eead9e0d8b : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"763f7b5b7147617f5327d4b72bebd63a2c6fa58e","revisions":{"587272239403b265017876c0c65615ae515abbbc":{"kind":"REWORK","_number":1,"created":"2019-10-25 13:33:32.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/08/691308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/691308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/691308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/691308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/691308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/691308/1"}}},"commit":{"parents":[{"commit":"0e8f37c35a9ef89d12d8cfcc507ed8c724badd98","subject":"Merge \"Image deletion returns 500 if \u0027file\u0027 store is not enabled\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e8f37c35a9ef89d12d8cfcc507ed8c724badd98"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-03-29 05:56:43.000000000","tz":0},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-10-25 13:23:03.000000000","tz":120},"subject":"Periodic jobs are failing for python 3.5","message":"Periodic jobs are failing for python 3.5\n\nPeriodic tips jobs are failing because python 3.5 interpreter\nis not available on testing nodes. It is decided that Stein\nbranch should be tested with python 3.6 version.\n\nRenamed periodic tips jobs to 3.6 and added python 3.6 enviornment\nin tox.ini to run periodic tips jobs against python 3.6\nenvironment.\n\nChange-Id: I8a763eff7c6524969b1fabf98ab0a5624954d8b3\nCloses-Bug: #1822238\n(cherry picked from commit 79e4adaf436a1e231db68361a85c5ca903ccda61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/587272239403b265017876c0c65615ae515abbbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/587272239403b265017876c0c65615ae515abbbc"}]},"branch":"refs/heads/stable/stein"},"763f7b5b7147617f5327d4b72bebd63a2c6fa58e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-20 11:28:42.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/08/691308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/691308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/691308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/691308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/691308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/691308/2"}}},"commit":{"parents":[{"commit":"621593f54da49a1b48f30f7c12e9803228dc2336","subject":"Bring py27 tests back","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/621593f54da49a1b48f30f7c12e9803228dc2336"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-03-29 05:56:43.000000000","tz":0},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-20 11:22:26.000000000","tz":60},"subject":"Periodic jobs are failing for python 3.5","message":"Periodic jobs are failing for python 3.5\n\nPeriodic tips jobs are failing because python 3.5 interpreter\nis not available on testing nodes. It is decided that Stein\nbranch should be tested with python 3.6 version.\n\nRenamed periodic tips jobs to 3.6 and added python 3.6 enviornment\nin tox.ini to run periodic tips jobs against python 3.6\nenvironment.\n\nChange-Id: I8a763eff7c6524969b1fabf98ab0a5624954d8b3\nCloses-Bug: #1822238\n(cherry picked from commit 79e4adaf436a1e231db68361a85c5ca903ccda61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/763f7b5b7147617f5327d4b72bebd63a2c6fa58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/763f7b5b7147617f5327d4b72bebd63a2c6fa58e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
