)]}'
{"id":"openstack%2Fglance~692034","triplet_id":"openstack%2Fglance~stable%2Ftrain~Ia76b107a137a39716253631e5a8d294f7ee9614e","project":"openstack/glance","branch":"stable/train","topic":"use-opensuse-15","hashtags":[],"change_id":"Ia76b107a137a39716253631e5a8d294f7ee9614e","subject":"Switch to opensuse-15 nodeset","status":"MERGED","created":"2019-10-30 06:09:13.000000000","updated":"2019-10-30 14:33:05.000000000","submitted":"2019-10-30 14:29:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692034-1572445743144-cd6aae0a","meta_rev_id":"e7f791d4142f145bbd03258da565c081f278d21b","_number":692034,"virtual_id_number":692034,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 14:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-10-30 12:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-10-30 13:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-10-30 13:02:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 12:39:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-30 13:02:25.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-10-30 14:29:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae3a8f2c67c28835a173d97a1ef2b47294108644","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-30 06:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4f7795d08fec24d9d3c2e162467c2d01c7928db","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-30 06:12:14.000000000","message":"Topic changed from train/backport/689018 to use-opensuse-15","accounts_in_message":[],"_revision_number":1},{"id":"790621e27c7b776122f1e2a0629e9c3c25d1797d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:43:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67b9715cb2394a0a8d56bb4f8ba4cc18 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98dd9b7201fe47cab8128b10a6585150 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b085301d6ffe4055ad42103f3faa10a8 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e667b6289214dcaa098324036a0a77a : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc0faac1b553410db1d9276469cab7ae : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c8bdd4ab724c859a39347af3903183 : SUCCESS in 6m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5978d3f673e74cfc97f8ed37c5687c06 : SUCCESS in 9m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/24eefe0b505248e9b9b35da9215eed28 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c7cb30c580205fe4765a290f9a0eec3ed5261a03","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 12:39:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7814b3ae02f2cbdc10d90007101c3a212dd135b2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-10-30 13:02:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMeets backport criteria.","accounts_in_message":[],"_revision_number":1},{"id":"f2036b3ea039f78bc34e2d4de057056f8fbd5822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:02:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40e82f547ab8fc0dfa2a0d1eb7ab555a368ca431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:29:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74750043ed04bd796ccf32a0fe67fd6 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ab55fb54bf41dcbb8a2e570a553c31 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae43fc0f48e041eeaca8461191a202fa : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8805a973b85d4c349e11d212cdf19b8f : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6561794b04274120886916721f9389ce : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c287ce639147738fec1b17570a7559 : SUCCESS in 8m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/763b587927ce40858990e77a9a9c4687 : SUCCESS in 11m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b0694c5a5d52442da288aa8f27d5aea3 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4b10be50b2f0d1dd0ee22217624ff5efded5414c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:29:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e7f791d4142f145bbd03258da565c081f278d21b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 14:33:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838373b928bb43e19fad902b7ecefc05 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e3ced8251cd6e273aa73f553a24fc475b219db5","revisions":{"6e3ced8251cd6e273aa73f553a24fc475b219db5":{"kind":"REWORK","_number":1,"created":"2019-10-30 06:09:13.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/692034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/692034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/692034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/692034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/692034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/692034/1"}}},"commit":{"parents":[{"commit":"f87464e022046c9d61384c2593edf9753ed34749","subject":"Train final release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f87464e022046c9d61384c2593edf9753ed34749"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:41:09.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-30 06:08:45.000000000","tz":60},"subject":"Switch to opensuse-15 nodeset","message":"Switch to opensuse-15 nodeset\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nThe new job tempest-full-py3-opensuse15 use the opensuse-15 nodeset,\nchange tempest-full-py3-opensuse150 to tempest-full-py3-opensuse15.\n\nChange-Id: Ia76b107a137a39716253631e5a8d294f7ee9614e\n(cherry picked from commit e45d5e3b9259620914ff3e3c5fa67dd23de1879c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e3ced8251cd6e273aa73f553a24fc475b219db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e3ced8251cd6e273aa73f553a24fc475b219db5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
