)]}'
{"id":"openstack%2Fglance~692043","triplet_id":"openstack%2Fglance~master~I8bb7ad9647c369655e611226ab2d3b4f86cfc41c","project":"openstack/glance","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I8bb7ad9647c369655e611226ab2d3b4f86cfc41c","subject":"Drop python 2.7 support and testing","status":"ABANDONED","created":"2019-10-30 06:42:35.000000000","updated":"2019-11-22 22:41:18.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"90664cb11591b277d711d3ec4a3eb8effdcb6cfc","_number":692043,"virtual_id_number":692043,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-10-30 08:28:52.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2019-10-31 06:31:40.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-30 10:20:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2019-10-30 12:42:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 08:28:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-30 10:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-30 12:42:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"2014e6ba2a57991a5668d8d398b6bb681bfdd9f0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-30 06:42:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df6245e4c5dcd9c5dcb4c140bdc76344eb0f6ae6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-30 08:28:52.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/43/692043/1/check-not-vote/ext-glance-zuul/282b6ab : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9ab8c81bb667ddffec6dc06a07b879183375a594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 10:20:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/50cffffcd5a3420a9ef55fa772bf29a8 : SUCCESS in 3m 43s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/47055c86db964567ab3916ad820b290e : SUCCESS in 6m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db992582f1bf48e38aac8793b36987a6 : SUCCESS in 1h 51m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/939cf898367d485f951d42b453a5ccba : SUCCESS in 1h 09m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7317d1283cbb4b60b1ec4b82efaa97e2 : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dab5d71eab34fb29a959a1976fc35e6 : SUCCESS in 1h 13m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dffce0b345de4236802dd4ad48cb8446 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff5c87456c2466a9a00b87c6c2e49ed : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1744a17904c74af080d39756bcb0846d : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccdf68ccb60341a783a534cde98a65a6 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d391fd0a344f46a264575162789c29 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4279c31dc2ae417d98dfae4eb77a0372 : SUCCESS in 9m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7a7f13761e604281961bcd194c41d37b : SUCCESS in 9m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b5c48b1af5594f458bce183ec979ab2d : SUCCESS in 10m 21s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/2c240027674f4147be19913fd2f4b4dd : FAILURE in 5m 09s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/32c0b79aa1e74b9f8608888800bca3f0 : SUCCESS in 4m 39s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d23ff4db6162417dba91f858b29742ca : SUCCESS in 1h 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd5ffb66d3d8bbfb3eed8a5ac4bbe17d57fb6fae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-30 12:42:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nNote, the concerns discussed in https://review.opendev.org/#/c/664477/2 still need to be resolved.","accounts_in_message":[],"_revision_number":1},{"id":"b253c7d9472e9417f1215b22925dc4823917eaa6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-31 06:31:40.000000000","message":"Patch Set 1:\n\n\u003e (2 comments)\n \u003e \n \u003e Note, the concerns discussed in https://review.opendev.org/#/c/664477/2\n \u003e still need to be resolved.\n\nI see, py36 and 3.7 jobs are running on glance so it should be fine to add the classifier also. Or glance is not ready for py3 yet ?","accounts_in_message":[],"_revision_number":1},{"id":"d255d3997b93435e65ee99567c0018e7e6c1e9f4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 22:41:18.000000000","message":"Abandoned\n\n@Ghanshyam: I stole your release note and made you a co-author of https://review.opendev.org/#/c/695707/ -- hope that\u0027s OK.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e50c320a569a9c4705d5058fa499e722ca3c80dd","revisions":{"e50c320a569a9c4705d5058fa499e722ca3c80dd":{"kind":"REWORK","_number":1,"created":"2019-10-30 06:42:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/43/692043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/692043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/692043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/692043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/692043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/692043/1"}}},"commit":{"parents":[{"commit":"e45d5e3b9259620914ff3e3c5fa67dd23de1879c","subject":"Switch to opensuse-15 nodeset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e45d5e3b9259620914ff3e3c5fa67dd23de1879c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 06:41:43.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-10-30 06:42:33.000000000","tz":0},"subject":"Drop python 2.7 support and testing","message":"Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nGlance is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: I8bb7ad9647c369655e611226ab2d3b4f86cfc41c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e50c320a569a9c4705d5058fa499e722ca3c80dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e50c320a569a9c4705d5058fa499e722ca3c80dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
