)]}'
{"id":"openstack%2Fglance~694238","triplet_id":"openstack%2Fglance~master~I77a57228d2e66558fc3a10f4a2d5adede9e87ffe","project":"openstack/glance","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I77a57228d2e66558fc3a10f4a2d5adede9e87ffe","subject":"Drop old neutron-grenade job","status":"ABANDONED","created":"2019-11-14 10:37:10.000000000","updated":"2019-12-11 19:56:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8284368107997ed0de9b50461de3f31db09f999","_number":694238,"virtual_id_number":694238,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-14 12:10:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2019-11-15 07:41:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-11-18 08:27:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2019-11-15 12:06:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 12:10:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-15 07:41:26.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-11-15 12:06:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-11-18 08:27:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0bb571c5cc506db1fab1fee944d697e19123da27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-14 10:37:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b601f0289f38636d6ebbee45b6ec40e663d0575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 12:10:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/de95f276541c43c983c81386553d93bd : SUCCESS in 1h 30m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc301068707f4f3f957b43b7d4262d56 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41e831c2f94428ebce168c5cbc8298d : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f52535604204445a6cfc58ce5d9f11e : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d590558c41a476ebea4ca8d4ace1299 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c2045deda6d4d78947d35d6c9dfdee5 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae90fd739eb94a68a5eace103bc6c6a1 : SUCCESS in 5m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d1973f09c3ec42fcadf70a1e93e0340e : SUCCESS in 12m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b517720c90ae43ecb03b6e3a754fee41 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4e6f175b9816e61ac683ef579c1c2cf162d64e6d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-14 12:14:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ec2d6a26810ae7e13afca5fa4ee62e18115bb22","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-11-15 07:41:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ab2947ae969c732206f588216ccdb23519935bb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 12:06:54.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e (1 comment)\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"eaf00e6e9185ac208913867b5bc0ab0ec6cef163","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 12:07:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"487c3d6458f63b471463e9cd96ffa0cccd2e723a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 12:28:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/818a4a754daf4270b0180025ba916786 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5635f874564baeaa559085e93f6a0e : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cf0ffcad09d434795f7f24482c25354 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69f28b585d4842abb7585ab69cbd9528 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98ec1ca84cc1465d950a609518b7a408 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aedcb98f0794b728a49a52de93d926b : SUCCESS in 5m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/06a24f254b0e40eb9d7f1675427f61df : SUCCESS in 10m 53s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1b2f62c263424deab1a8c76a34a24126 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":1},{"id":"855c301d0093acea3bef5c2cc37f11de2b886850","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-18 08:27:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f17cae30cec07eaa666003e5e6860e5cce5f124d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-11 19:56:26.000000000","message":"Abandoned\n\nit\u0027s not needed as this job was removed with https://review.opendev.org/#/c/698410/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"091c6b805de7a96469b67795b7ae1c1b21f7ddaa","revisions":{"091c6b805de7a96469b67795b7ae1c1b21f7ddaa":{"kind":"REWORK","_number":1,"created":"2019-11-14 10:37:10.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/38/694238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/694238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/694238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/694238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/694238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/694238/1"}}},"commit":{"parents":[{"commit":"e45d5e3b9259620914ff3e3c5fa67dd23de1879c","subject":"Switch to opensuse-15 nodeset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e45d5e3b9259620914ff3e3c5fa67dd23de1879c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-14 10:36:37.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-14 10:37:06.000000000","tz":60},"subject":"Drop old neutron-grenade job","message":"Drop old neutron-grenade job\n\nThis job is still running python 2.7. As we are dropping py2 support in\nUssuri cycle, lets drop this job now.\nThere is same job called \"grenade-py3\" which runs on python 3 already\nand this is still used in project\u0027s CI.\n\nChange-Id: I77a57228d2e66558fc3a10f4a2d5adede9e87ffe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/091c6b805de7a96469b67795b7ae1c1b21f7ddaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/091c6b805de7a96469b67795b7ae1c1b21f7ddaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
