)]}'
{"id":"openstack%2Fglance~694387","triplet_id":"openstack%2Fglance~master~I33b84c4d68e8077271447bcbdea4b7052eb01204","project":"openstack/glance","branch":"master","topic":"694387","hashtags":[],"change_id":"I33b84c4d68e8077271447bcbdea4b7052eb01204","subject":"Remove empty policy.json","status":"MERGED","created":"2019-11-14 19:08:00.000000000","updated":"2020-03-05 17:50:11.000000000","submitted":"2020-03-05 17:40:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694387-1583430036810-eb552905","meta_rev_id":"bf67c12c11b9f476b3ae1811589e4af0034fe89a","_number":694387,"virtual_id_number":694387,"owner":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"date":"2020-01-16 22:44:28.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 17:40:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-05 15:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-02-06 08:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-05 15:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-02-06 08:58:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 00:31:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-16 22:44:28.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-06 08:58:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-03-05 15:50:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-05 17:40:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"551bb36fb9d8b7bdc40540e936f36e6cf5639b6e","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-11-14 19:08:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52ad363c20e7253066c97fb82f6d0f9c528c99b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:52:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/242e371e87a04e69bc8392be6ddfd163 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9bc91387a1f444b936f6c5971e10e82 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04b65b41f6674429bc6d63397191439f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67e171db4c0e4684bd71bdfd4d2bf32d : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b2a8b88a54343928c417dde28a1503e : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6a699ef13f4d768ba6244dc82b5b5f : SUCCESS in 5m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d2fbafe165a54c8e9ff44da68681bc13 : FAILURE in 9m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a0c475f87a264540a3d660e559162f67 : FAILURE in 9m 45s","accounts_in_message":[],"_revision_number":1},{"id":"43b5c2f145afb0aaf8bb48c383c65ba500640f17","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-11-14 21:40:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b17ec7706f6b2ad97da57720ce729a26539083d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 21:56:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/697661ced51c4ffa817993e946a74576 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12b9c4cffb24138ad86d8407f590c72 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b83a606cd575427da6a8eb4559fb881f : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f503f654bc844d608a1acd7ce2f054f1 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d478b3c7cc040f9b989a0c60061722b : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32326cd304184154b7766bea294e5a05 : SUCCESS in 5m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e12bf7179ec3461d8f1814f318642739 : SUCCESS in 9m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2dd3410f53254f3f9072cb2d24aba7c4 : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4241e545e4721cc62718edf28e2c3aa477fce3b8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-20 10:18:26.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/87/694387/2/check-not-vote/ext-glance-zuul/e041666 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":2},{"id":"07a995b14609caefd38e3057f6f610ddcaba7c7a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-04 13:07:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"462ef86ea163c694ead98e2d88d34490c88c5096","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-05 15:00:33.000000000","message":"Patch Set 2: Code-Review-1\n\nI think this requires a release note and a documentation update.  In particular, we need to make sure the property-protections docs are accurate.  Maybe the property-protections update can be a follow-up.","accounts_in_message":[],"_revision_number":2},{"id":"fe00b1d7b5b05c90a947e24923976547886aadee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-05 15:01:58.000000000","message":"Patch Set 2: -Code-Review\n\nGood point. At least a release note making it clear for upgrades that something here has changed.","accounts_in_message":[],"_revision_number":2},{"id":"9dc9cd6a9f96927fb4434021358cf9689e2edd4a","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-05 22:27:16.000000000","message":"Patch Set 2:\n\nDo y\u0027all mean something different from what is in the previous patch?\nhttps://review.opendev.org/#/c/693129/9/releasenotes/notes/policy-in-code-7e0c6c070d32d136.yaml","accounts_in_message":[],"_revision_number":2},{"id":"58d2dc52a3f5dce7e78b044ad4d454626099ef17","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-12 02:56:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6dd5912233a373aac85dcc706a4d27cfaea39f3c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-12 02:56:58.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/87/694387/3/check-not-vote/ext-glance-zuul/44f5889 : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"debb5aabb5eccac74668e7a737f617d352e8c0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 03:10:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/158babac63bc4e629c224d22c5f507a3 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196c9d930a67422fa087c424a8be9ae4 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e79f0faf817940f3ad86a2b62a76e333 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b1ebd4ac4a48aa993debde148003e3 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98e4ed57fcf54fa3bb44e62492168e24 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfea11b8d0a747709a980170586f1750 : SUCCESS in 5m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a782bb390e87418f83588057c161153e : SUCCESS in 12m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/11d16e1cfa9f498da5aa494ef43b7ec2 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":3},{"id":"e0beb46a0675f488fbaa7f917bbc728248db8006","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-12 19:45:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"df95bd6794749f9c9823e9b74dc0d6dec19ee468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 21:46:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca37617e97d84982859e65f878c51c46 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced30f0a16b140f185d0ecef3626b27d : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7b6e5e3b6bb4468bd25b98f63c89d9e : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a17593509b514cc49dbee8f70a2e874b : SUCCESS in 11m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b29915b55ba846eb8410a49d57739afe : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197dad15c8134e9691e67fbc30ea0694 : SUCCESS in 11m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e68bbb954df468781eb8b9c0b6976df : SUCCESS in 13m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/72c90d04971b40bbb37feb81b6f1e37e : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":4},{"id":"b9a124c2b5c2953bd35045771809399022e894e9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-13 07:05:22.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/87/694387/4/check-not-vote/ext-glance-zuul/00570e4 : FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"7d7f1d5a487d5a414d53f0e76ca02f6251b608fd","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-16 14:33:59.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"925804ba06e236b91bf189cac2128cf4e74968a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:50:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c67040045594bc1b8516fbf7b37a516 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544f6655b2cf4401816a1d352544a260 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb26020e05ce400c91a32109f4401045 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2941c9b1381042f0915a2dff2f5d5564 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/223a198652c84cc88a67c60d0c1d7fa1 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcaacba5b5e43bcbcca24f6f4d62ce3 : SUCCESS in 7m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/325a7f61fffa4c48979e55dcd8266343 : SUCCESS in 12m 17s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/345db5e3ee9a489bb8db369300e10d76 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":5},{"id":"234fe223c274cc2c7db59bb08b41f7ede00fe3c6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-16 16:21:40.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/87/694387/5/check-not-vote/ext-glance-zuul/9803a1a : FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"032fba60373f690063a4aca6a2cbaf9da4c9c58c","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-01-06 18:01:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"4a19cfa56d7f8280181ac8cb38cf168b8c5284fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:28:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23fc7b2dabee4886a4c0cd7c2699ccfd : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a510e35f18a4cdc92404e2c177a3263 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b94c3ad92ce4c59acbda4d37c8afe33 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/142ad54f7e9041c9b7b8ab772597a195 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf8bb2154d04b60a7abe61f723b4d72 : SUCCESS in 9m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f25b9dced044c1ea0089ce851c1bccd : SUCCESS in 13m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7493f8e78c5242c7bc781574890246b5 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":6},{"id":"81d572ea7ac6926a4f373d60639f3131f0f6798a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-07 00:31:30.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/87/694387/6/check-not-vote/ext-glance-zuul/3e03964 : FAILURE in 1m 32s","accounts_in_message":[],"_revision_number":6},{"id":"573b624188f3cd960cf7184d570a73b91101b2bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-16 22:15:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"499a966bca706774d7fefb6c548b856dd4a6f297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 22:32:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4e85872e4f04819bfa3085c111cfc02 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92feeeadaa9542429a39b83faa9d5913 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/710a78b770d14f9b9d2377ca1b7c76ca : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65ba8c7f85d545dca5aa51dea535f380 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4669873fb0644d39032bbaafacd865b : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be0092df45c4bad80a2456cf0ae0229 : SUCCESS in 11m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8e5716edc0e45efbe7b6cd6c45e2daf : SUCCESS in 11m 53s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a2fed2f3ac3e4f2bba5fc41229ea1147 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":7},{"id":"96e018cfa890953d9a5352de3346208c6caf62c7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-16 22:40:24.000000000","message":"Patch Set 7:\n\nI think the property-protections docs are OK, but the policy doc is way out of date.","accounts_in_message":[],"_revision_number":7},{"id":"f22db0e56b1bcccdb4f48ac6318f313cc51b2b66","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-16 22:44:28.000000000","message":"Patch Set 7:\n\nI think we should add something like https://opendev.org/openstack/cinder/src/branch/master/etc/cinder/README-policy.generate.md so that there are instructions somewhere for how to generate the file.","accounts_in_message":[],"_revision_number":7},{"id":"0c762b5909e87efc9d8b9f216833d706d990df8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-06 08:58:32.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"7cebecbb762935e4f92e5ffb6f437aa5d4e6ec22","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-05 15:50:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"fdaf3ff6e2820ffb3f6948e59202b51e5d03752c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 15:51:07.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"833f2c2006c73792f8e863a62ada992d38501957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:40:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f26ca6e66461ce87bb7b294d817d8f8ebeb9001a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:40:36.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eff3b680bcc7448299099764989fc5bc : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fab5028724146ed805431a898c1a840 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781559977be24a558b53c52ffe903eae : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d30028ef594421d81e68d6e45e2d937 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798f099d363142c98e73cd55dfc6c0b2 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10b2d66739324f6899f5be2967734f6d : SUCCESS in 10m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fe8f492243442eb9662d1954cd5ce99 : SUCCESS in 7m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/827348668c004aa299d9ecfb1728a5fe : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":7},{"id":"91e34e3f8e1ff43dc180623a87b262f5c4e0bd2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:50:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d35031573ca40c2939b149d7b4060f8 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e0f6d033c16447d9bba8e7b6e0e6c30 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5bffac018656145a986a9a8227167cf0e1890be8","revisions":{"e22c3ee4994786e7661ebb4d91d67605a0d61fac":{"kind":"REWORK","_number":1,"created":"2019-11-14 19:08:00.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/87/694387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/1"}}},"commit":{"parents":[{"commit":"99dff279aec745f0d8b18d9d6fe5946f27da60cc","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99dff279aec745f0d8b18d9d6fe5946f27da60cc"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:07:11.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e22c3ee4994786e7661ebb4d91d67605a0d61fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e22c3ee4994786e7661ebb4d91d67605a0d61fac"}]},"branch":"refs/heads/master"},"199b12b81c224e61f6c1aba0bbe482d4f2c7b9f7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-14 21:40:22.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/87/694387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/2"}}},"commit":{"parents":[{"commit":"649e06c58f4e860cd5167966ed87208a2f76003f","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/649e06c58f4e860cd5167966ed87208a2f76003f"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 21:39:05.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199b12b81c224e61f6c1aba0bbe482d4f2c7b9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/199b12b81c224e61f6c1aba0bbe482d4f2c7b9f7"}]},"branch":"refs/heads/master"},"f162f2e260776e9d82fd0424efabb1b96228cca7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-12 02:56:28.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/87/694387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/3"}}},"commit":{"parents":[{"commit":"9e8a7bd7069d45f9f7e2de8dc91dd186e29ed08f","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e8a7bd7069d45f9f7e2de8dc91dd186e29ed08f"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 02:56:05.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f162f2e260776e9d82fd0424efabb1b96228cca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f162f2e260776e9d82fd0424efabb1b96228cca7"}]},"branch":"refs/heads/master"},"029ccf17a0a80cd1272752b901e9f7dbdb19be4c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-12 19:45:01.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/87/694387/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/4"}}},"commit":{"parents":[{"commit":"9860e20083e3e59f4a6647308e20d63f6d76762b","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9860e20083e3e59f4a6647308e20d63f6d76762b"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 19:34:59.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/029ccf17a0a80cd1272752b901e9f7dbdb19be4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/029ccf17a0a80cd1272752b901e9f7dbdb19be4c"}]},"branch":"refs/heads/master"},"89e2276f670f77239ad10842861ad3e6110b5d4a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-16 14:33:59.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/87/694387/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/5"}}},"commit":{"parents":[{"commit":"37f8926810cab176620235425c9f45fb71c0cfd1","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37f8926810cab176620235425c9f45fb71c0cfd1"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-16 14:33:43.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89e2276f670f77239ad10842861ad3e6110b5d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89e2276f670f77239ad10842861ad3e6110b5d4a"}]},"branch":"refs/heads/master"},"f4520756a1645171e9040f6f3692bd907e667c44":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-01-06 18:01:52.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/87/694387/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/6"}}},"commit":{"parents":[{"commit":"5c17e4c7ef533b57ba15ead931b0452e23f46b4b","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c17e4c7ef533b57ba15ead931b0452e23f46b4b"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2020-01-06 17:56:30.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4520756a1645171e9040f6f3692bd907e667c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4520756a1645171e9040f6f3692bd907e667c44"}]},"branch":"refs/heads/master"},"5bffac018656145a986a9a8227167cf0e1890be8":{"kind":"REWORK","_number":7,"created":"2020-01-16 22:15:49.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/694387/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/694387/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/694387/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/694387/7"}}},"commit":{"parents":[{"commit":"5c17e4c7ef533b57ba15ead931b0452e23f46b4b","subject":"Move policy defaults into code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c17e4c7ef533b57ba15ead931b0452e23f46b4b"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-11-14 19:05:38.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-01-16 22:15:36.000000000","tz":-300},"subject":"Remove empty policy.json","message":"Remove empty policy.json\n\nSupplying a policy.json file is no longer necessary.\n\nChange-Id: I33b84c4d68e8077271447bcbdea4b7052eb01204\nDepends-On: https://review.opendev.org/694386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5bffac018656145a986a9a8227167cf0e1890be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5bffac018656145a986a9a8227167cf0e1890be8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
