)]}'
{"id":"openstack%2Fglance~694723","triplet_id":"openstack%2Fglance~master~Ib10304b9f65fb26b2b953062ad771806728265da","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ib10304b9f65fb26b2b953062ad771806728265da","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-11-18 07:54:26.000000000","updated":"2019-11-19 10:36:18.000000000","submitted":"2019-11-19 10:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694723-1574159667749-ded4c5e8","meta_rev_id":"5b92d7f96fb15fdee4898261dde5567f0a0f0340","_number":694723,"virtual_id_number":694723,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 10:34:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-19 10:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-11-19 10:18:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 10:18:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-19 10:34:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a90aef3ff4852cbe4d08108975af59443fe78d63","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-18 07:54:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61b045bc32aa5d683d860a30d342f5943122ebaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 08:06:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7547890cfe174a4dac23037232f95821 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86041f4894ea417ba10105adef956156 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2255f2bdd694e4096ae89eb687cef87 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb3fc60e7f34d60a8f9b78b055546dd : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/595cdeef2c7e4c8db1392bffd526e0a8 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0177f6b6e14348109246641a27652c84 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfeef15a69654831ac6d8e6ecc5ab2a4 : SUCCESS in 9m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/32016c2297a64b5cbc91e66cf3e83380 : SUCCESS in 9m 58s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f53f3cfb5dbb4f74b8f032f6d13a85d2 : SUCCESS in 9m 35s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/63ddda3a68484979afacb7e6e77c0b58 : FAILURE in 4m 27s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b8aa94168898438e886fa66350e7afb2 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"84118f0bff0c46d42c12b4f49295599a46d795f1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-18 14:54:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82bb160ee9426e0117a7b07add64ab6a8c58a548","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-11-19 07:46:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c83328449d77b19656d56dac3f88f8cfc65a65be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 08:58:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cceb9e5b47254a13ae61392c4b495756 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09b32d6b86c43b4b13ad9255c7367f6 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/754ed0fc28d04cda9463b8e292cb0490 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5024f2748bac45dc89d0c41234166c99 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da6d3542a2fd412081c0b26d23fd9fa0 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ab59b4f8394d839e5dd9fb372ec147 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/130f5ec0994d40d68ef69cd15f56271b : SUCCESS in 9m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2a831ddd2386424d973deed0dac8e0b3 : SUCCESS in 9m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a7e5e1f25062452a92d8d9183bbb903d : SUCCESS in 9m 44s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/2ab20eed8e7241778d5b0564b709c4e8 : FAILURE in 5m 17s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/382bea701fc3439ca4a5b8897f2117a3 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"aca0b59e4c2378bacca23ec6263e3e8d5d7fbbe8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-19 10:18:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"234ae517b07001877c76276999e87e9291b687d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:18:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81c61c1c8da0e8e20be300d8f304336aacb14950","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:34:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5b62b9375ddd04fae4704a2dbd4712083ea9f01f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:34:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/233e8f314a0b4ec39f8f00e69aecc2e1 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b6c1208696411a82cbb85fa5b7e300 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/902ec8ce51d94f459d54c92d1884f50f : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/796f626c41364de6897aa6f96912e981 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3636dbfd9c094b7188cebd3f7568a101 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b8450a69284a03bf71b34c63001187 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a69702213c84f0e88783f5c90176fe9 : SUCCESS in 9m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/34988b773cf14058a6747056cac4c6e4 : SUCCESS in 8m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fc6b40cfad264eeb9e971ea11ec59683 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0ed7e9cdac75b450942b9facb3a7299160cae0a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 10:36:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b35d8b254849cb9e5dc34c4bb9aec7 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8e7bf9c36e642eca786bb824b325637 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d547e277ad295e870943a73a2319a82ef8f4ef6","revisions":{"519f0814f812efae10218b5a2a50c3f3e6d5ca4e":{"kind":"REWORK","_number":1,"created":"2019-11-18 07:54:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/694723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/23/694723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/694723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/694723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/694723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/23/694723/1"}}},"commit":{"parents":[{"commit":"e45d5e3b9259620914ff3e3c5fa67dd23de1879c","subject":"Switch to opensuse-15 nodeset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e45d5e3b9259620914ff3e3c5fa67dd23de1879c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-18 07:54:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-18 07:54:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib10304b9f65fb26b2b953062ad771806728265da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/519f0814f812efae10218b5a2a50c3f3e6d5ca4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/519f0814f812efae10218b5a2a50c3f3e6d5ca4e"}]},"branch":"refs/heads/master"},"4d547e277ad295e870943a73a2319a82ef8f4ef6":{"kind":"REWORK","_number":2,"created":"2019-11-19 07:46:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/694723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/23/694723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/694723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/694723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/694723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/23/694723/2"}}},"commit":{"parents":[{"commit":"30c9275888e1f598c95f556eeeccb120efca3722","subject":"Merge \"Start README.rst with a better title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30c9275888e1f598c95f556eeeccb120efca3722"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-19 07:46:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-11-19 07:46:48.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib10304b9f65fb26b2b953062ad771806728265da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d547e277ad295e870943a73a2319a82ef8f4ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d547e277ad295e870943a73a2319a82ef8f4ef6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
