)]}'
{"id":"openstack%2Fglance~695102","triplet_id":"openstack%2Fglance~master~I6216868fc5405cc3f3a14cd5f32a38d778e64c2d","project":"openstack/glance","branch":"master","topic":"drop-py27-support-devstack-default-py3","hashtags":[],"change_id":"I6216868fc5405cc3f3a14cd5f32a38d778e64c2d","subject":"Migrate grenade jobs to py3","status":"MERGED","created":"2019-11-19 22:24:22.000000000","updated":"2019-11-21 16:03:50.000000000","submitted":"2019-11-21 16:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695102-1574352142471-ded4c5e8","meta_rev_id":"6fc2a7d750d0de522976f964300ce6ae8e61d78a","_number":695102,"virtual_id_number":695102,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 16:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-19 23:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-21 14:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-11-20 09:05:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-21 14:32:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 23:26:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-20 09:05:16.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-11-21 14:32:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-11-21 16:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0ef2b210b117e0d4eb3153605f397ac60efc38a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-19 22:24:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"290ef2b64e56786f7dc7e7cb0682b27b208a4135","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-19 23:03:55.000000000","message":"Topic changed from drop-py27-support to drop-py27-support-devstack-default-py3","accounts_in_message":[],"_revision_number":1},{"id":"7c49fcfa65ca11925e3595c14d604f129acf9841","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-19 23:26:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33451a2d4a06c22399e2b131a4d215068406a3d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 00:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0dc7d8eac38149eaaee03c55cafe03e2 : SUCCESS in 1h 35m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fdab52468d2945b6a484949a4e91fc78 : SUCCESS in 1h 06m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c1772ffb879142a99e5cc5cdeda81dfc : SUCCESS in 56m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fc0352fb8b04c11b7fa28a4cd7038ef : SUCCESS in 1h 23m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c91bf0dd9b547a8804984ab11abbc52 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4531fb6bcc480aac9639a91c8307fc : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a66d1dfadc7d4ea582e77c327e662e3d : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7c635a69084c0a97b121dd088cfc9d : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e04a5f90d994390b44357e9027c2dc2 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b47a1f008354eac9bd3d5dfec2fd9ce : SUCCESS in 5m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3c858229f77e4dac9e81fe9d89c4d931 : SUCCESS in 8m 52s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2bb3d472c8a84873993e8bc8db2cad95 : SUCCESS in 9m 29s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/a67e906d30a84897856b3684b0258140 : FAILURE in 4m 25s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/890d5b1d43544ef9a42e48c3e15daea1 : SUCCESS in 4m 33s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c014a73ed8a4d97bb87ef530d6844a4 : SUCCESS in 1h 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d567d61df130b2db4022f019a75d6f69cdf771c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-20 09:05:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f5e50d51e3f8f48cc3c1eb70db29651734ea37a8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-21 14:32:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"31731ab84f5b516933150e2b6ceab5795360949e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 14:32:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43e501190eb27a432045e148e02e3fb9ee840dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:02:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bed6bfac4aa044a48f0802ff0d9f731b : SUCCESS in 1h 17m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2bbfd59e680b4037abd4db4e758417d6 : SUCCESS in 55m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c4e3e699de44150b5f349ffb151d60c : SUCCESS in 1h 08m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4254607fe2e402ebcd67017844d12c6 : SUCCESS in 1h 18m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6dd9312987340d3862ab360a0b6149a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33965fabd22a4d9d95102134a485b9e5 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85151c74f4a04490a371882c5a847b83 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b47efefee794e88ae37930fadfae6c1 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c859bc1b51a4a169f5acdf848f8e8c2 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d351f0500c594a97b5447c9c2d38d991 : SUCCESS in 7m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/31ab8147fdab4c84b23adffc41525b19 : SUCCESS in 8m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a14ec6712cd34213a2f292038a238f30 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1ed3a0906346c066b9aeb0ec1590207f03f957f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:02:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6fc2a7d750d0de522976f964300ce6ae8e61d78a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cc56a9ae30946c6926e65bb570ec632 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d23d12a46257b5c22c5765a85b89da7748155901","revisions":{"d23d12a46257b5c22c5765a85b89da7748155901":{"kind":"REWORK","_number":1,"created":"2019-11-19 22:24:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/695102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/695102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/695102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/695102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/695102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/695102/1"}}},"commit":{"parents":[{"commit":"a65462687718a731359d31bd14dc6756994d9b7b","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a65462687718a731359d31bd14dc6756994d9b7b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-19 22:24:01.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-11-19 22:24:01.000000000","tz":0},"subject":"Migrate grenade jobs to py3","message":"Migrate grenade jobs to py3\n\nAs part of community goal of dropping py27 support[1], we are\nmoving the devstack to py3 by default[2]. That will make grenade job\nto perform upgrade from py2 to py3 which will not work (we have seen\nthe failure in neutron-grenade job).\n\nTo avoid existing grenade py2 job break, this commit moves grenade\njobs to py3 which is what we planned as part dropping the py2 support.\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[2] https://review.opendev.org/#/c/649097/12\n    http://lists.openstack.org/pipermail/openstack-discuss/2019-November/010938.html\n\nChange-Id: I6216868fc5405cc3f3a14cd5f32a38d778e64c2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
