)]}'
{"id":"openstack%2Fglance~695177","triplet_id":"openstack%2Fglance~stable%2Fstein~I2dee048b71e3758208fae438423cceae661c8f00","project":"openstack/glance","branch":"stable/stein","hashtags":[],"change_id":"I2dee048b71e3758208fae438423cceae661c8f00","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-11-20 11:28:42.000000000","updated":"2019-11-26 09:55:14.000000000","submitted":"2019-11-26 09:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695177-1574761950136-ded4c5e8","meta_rev_id":"6e30173fbb2b32ef1e13afdb95c94e3b80648880","_number":695177,"virtual_id_number":695177,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 09:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2019-11-22 14:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-11-26 05:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-11-26 05:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2019-11-26 09:55:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2019-11-22 14:21:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-20 11:28:42.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-11-21 14:30:27.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2019-11-22 14:21:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-11-26 05:51:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-11-26 09:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c824aff6c0afb3d54f4e2e1778f3cf80e23e1d49","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-20 11:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3ea500dbcc30392c0a9f21c3cf0933ec9db1e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 11:29:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch stable/stein.  The error was:\n\n  The project template \"openstack-python3-stein-jobs\" was not found.\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate\n        - integrated-gate-py3\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-stein-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py36\n          - glance-eventlet-ssl-handshake-broken-py3:\n              voting: false\n          - devstack-plugin-ceph-tempest:\n              voting: false\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n      gate:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py36\n          - tempest-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - neutron-grenade:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n      experimental:\n        jobs:\n          - barbican-simple-crypto-devstack-tempest\n          - glance-dsvm-grenade-multinode\n          - devstack-plugin-ceph-tempest-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-full-py3-opensuse15:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^releasenotes/.*$\n                - ^setup.cfg$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py27-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py27-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py27-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@stable/stein\", line 200, column 3","accounts_in_message":[],"_revision_number":1},{"id":"be8c1903c9cdae44f50d54482a03debbfb104a3a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-20 11:48:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98e8e76320d14d10eaefb31432f37ae35a6fb655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:43:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/208d26db3b2e42d0b66da613d900c442 : SUCCESS in 53m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1b3a15e487e42e3aa3063b9979a6036 : SUCCESS in 1h 20m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32d157eae6f74d2ebb18c2a3ee5ec921 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df30d0802b894a92813bdbb8a6c8a037 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0d5957591774bb78f8266fa2192740b : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5ee1d25571840e2a340ddb77285b80d : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a17d6cd69f249c289fc1d5a10ee9c09 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7602844e4b4b48817c664783e9ac9b : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adedd0d6b4194cd08256c5eda5a18eb1 : SUCCESS in 9m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/298aae0e591047b1b4d73f3d0b466a10 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24b24aaae6d2472cbf633555e5f75418 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c35ab5db894e6ba8373e5795aaa6f6a11d282619","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-11-21 14:30:27.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0e3b872b0aeef822a736fccefd549fce3b2dd6d6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-22 09:25:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f1357cdb3a1e3f8b0161e39b09ba588381a9ed16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 10:34:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d24f7fd3792147788e25904b85c29749 : SUCCESS in 48m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e377303e73743a2873439a1bf7a73b1 : SUCCESS in 1h 08m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7dabb1569df49e4bc6e8be613ce13ba : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fdf497d1db463ca2e92e229b266860 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07201519e6da4cf39ce206addf14fc71 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3439616d5feb4e76a92713f5ed058b9e : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f578e1a02ddb4cbb8e46e9614a01fcfa : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04ec46d50464b7683db3bfaffa9a432 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75e11521a0fb45f3ad660ab162f668d9 : SUCCESS in 8m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/262f84ef76c24844b9407dde4485216a : SUCCESS in 9m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2585a1e4c80b4ace9eeed10b404b79bd : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d6fbf18617fbce037b2d3b49eec3b5ae53d38bc0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 14:21:21.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\nLooks like an appropriate change to keep the stable/stein check/gate operable.","accounts_in_message":[],"_revision_number":3},{"id":"43b3b2cc8ee6fc425c4c575bab43101f62fc21ce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-26 05:51:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you Luigi!!","accounts_in_message":[],"_revision_number":3},{"id":"b5bd397de4f5980d1b16c1956f07e16675592993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 05:51:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5822832bacbab6e94af0a0452a395284e98387e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 07:10:59.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c97baf52cba34bbaa29b78ce5f769a39 : SUCCESS in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9452b5a285a0429681dc711098b0913d : FAILURE in 1h 17m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54ffe562e985455c91d7ba54e42887f2 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e2eb047f4b436e9cf8f6b21edb4876 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6dedd4bec44548edaa497af6c9047959 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec0a22843dab4e82ac1ebff98a96d28e : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d190704112ec4b46bee8e75e12a24e89 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddd2ffe202a495e94ad3fe1d93a5c3e : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d32047f68a142b192466432038bc186 : SUCCESS in 10m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fb3f4c4daefc4ace8269f0be21b3a62c : SUCCESS in 9m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e86daa374ab49f484a3daab32c6f6d7 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":3},{"id":"92614169a52d90e678c27090ee8dc8babe6f81ba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-26 07:17:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb39e58f25e28cd7956575e78155089bbbb69ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 08:32:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0baaa0639aff418a925f653a1e032d40 : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/584740014aef462a951b0fe0c3a2b536 : SUCCESS in 1h 14m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee243b326f3545f6adb11e7d296c0adf : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa5cf633f5f4f34a5760cbd2f36f7eb : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/702dfd2963984bcea5ef68603b7012d0 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/facdd307cdfd42649658e149f169b6a7 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/051cdc9a10f7487b95d26f32104a0957 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b851a9a0b7a4d64b54fa79fe5088a94 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/142f71b12a8a4b118b7e34c860691c70 : SUCCESS in 10m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0d07bb68d2f0455f97b7112b173a9732 : SUCCESS in 10m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/16b6b6ddc2eb474abbc012368db8bccd : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4f579ed62f8e061ae4ddef4370a9c9499c796a1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 08:32:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6d6c936118a4ad02df3ddb16eb778ea116050358","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 09:52:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ff56557997a4750bdc361aa5e717a22 : SUCCESS in 58m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/246feea4d17249b98ee60ab4316bc887 : SUCCESS in 1h 14m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91f980b576c64e45be279736a7e3e7bc : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100c3e7f164e4e5bb7e97efa8d631f62 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d666fee0aaf4119ae9684044c725c0e : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3e1cdb035946aaa52ae3351b016a81 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b59ec46fa50461a81beff1142b6cee6 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44fa314868e34b29a5959b8ecd41f095 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b92a22069a4b2ab290f77934300309 : SUCCESS in 9m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40c7213d217b483f9cae3e29039f1694 : SUCCESS in 8m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ee1cb8e225a4790b46f09def8daa7ec : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":3},{"id":"abacc5fa1f3cadb6cbf62a1416cf8442382318fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 09:52:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ad29f8374c4dff5138feef21729aa152d6b918b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 09:55:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1ee157d3da4056a96a4ef5eedf0dc7 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b12bc95cd56445f91a47863df6b7f63 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a6aa296225048c41f68a39b984e81f19db2c5a20","revisions":{"010e7d245e7e98ef7c55315fc23b9cf8b8970eba":{"kind":"REWORK","_number":1,"created":"2019-11-20 11:28:42.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/77/695177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/695177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/695177/1"}}},"commit":{"parents":[{"commit":"cfcb050093e1d1e45dc034054fb884fbce075977","subject":"Don\u0027t target broken ssl tests to specific py3.minor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfcb050093e1d1e45dc034054fb884fbce075977"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-20 11:28:01.000000000","tz":60},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nStory: #2005924\nTask: #34208\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n(cherry picked from commit b7e9db70465f19d4a4b008a93147a0649ca36535)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/010e7d245e7e98ef7c55315fc23b9cf8b8970eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/010e7d245e7e98ef7c55315fc23b9cf8b8970eba"}]},"branch":"refs/heads/stable/stein"},"e27af228ca6a396ecb4cccd9547884720e253efb":{"kind":"REWORK","_number":2,"created":"2019-11-20 11:48:10.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/77/695177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/695177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/695177/2"}}},"commit":{"parents":[{"commit":"cfcb050093e1d1e45dc034054fb884fbce075977","subject":"Don\u0027t target broken ssl tests to specific py3.minor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfcb050093e1d1e45dc034054fb884fbce075977"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-11 01:08:53.000000000","tz":480},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-20 11:47:57.000000000","tz":60},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nStory: #2005924\nTask: #34208\n\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n(cherry picked from commit b7e9db70465f19d4a4b008a93147a0649ca36535)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e27af228ca6a396ecb4cccd9547884720e253efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e27af228ca6a396ecb4cccd9547884720e253efb"}]},"branch":"refs/heads/stable/stein"},"a6aa296225048c41f68a39b984e81f19db2c5a20":{"kind":"REWORK","_number":3,"created":"2019-11-22 09:25:43.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/77/695177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/695177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/695177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/695177/3"}}},"commit":{"parents":[{"commit":"cfcb050093e1d1e45dc034054fb884fbce075977","subject":"Don\u0027t target broken ssl tests to specific py3.minor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfcb050093e1d1e45dc034054fb884fbce075977"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-11-22 09:24:31.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-11-22 09:24:31.000000000","tz":60},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nStory: #2005924\nTask: #34208\n\nCherry-pick notes (stein):\n- keep python 3.6.\nChange-Id: I2dee048b71e3758208fae438423cceae661c8f00\n(cherry picked from commit b7e9db70465f19d4a4b008a93147a0649ca36535)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6aa296225048c41f68a39b984e81f19db2c5a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6aa296225048c41f68a39b984e81f19db2c5a20"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
