)]}'
{"id":"openstack%2Fglance~695707","triplet_id":"openstack%2Fglance~master~I15467c34fc8466fa1724c5401dc879ae77c1922d","project":"openstack/glance","branch":"master","topic":"bug/1853693","hashtags":[],"change_id":"I15467c34fc8466fa1724c5401dc879ae77c1922d","subject":"Remove py2 testing, jobs","status":"MERGED","created":"2019-11-22 15:15:40.000000000","updated":"2019-12-18 13:42:54.000000000","submitted":"2019-12-18 13:37:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695707-1576676252983-0548d2f5","meta_rev_id":"be133353fe611d708a9a527bbee44f43ef9b26dc","_number":695707,"virtual_id_number":695707,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-17 19:27:02.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 13:37:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-17 19:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2019-12-18 05:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2019-12-17 20:28:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-12-18 05:55:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2019-12-18 13:42:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 19:27:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 19:49:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-17 20:28:17.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-12-18 05:55:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2019-12-18 13:37:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4860bf85e39c5c55e7495fa9e00379435da29a62","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 15:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62f1210ad074f05a9726f27cad9be6eee77d529f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 15:16:49.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"d2ba114cf8b774fde3e090b338352565f642fd3f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 15:33:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- glance-tox-functional-py37-oslo-tips https://zuul.opendev.org/t/openstack/build/1db87fbd3c024c928087d17f2024447f : SUCCESS in 4m 19s\n- glance-tox-functional-py36-oslo-tips https://zuul.opendev.org/t/openstack/build/5a895ca1747a468dbc72abda494cdcd5 : SUCCESS in 10m 58s\n- glance-tox-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/c30aff19d347475ca1facae608608420 : SUCCESS in 4m 43s\n- glance-tox-py36-keystone-tips https://zuul.opendev.org/t/openstack/build/60ad3c937bf74edcb5807fd318bc1dda : SUCCESS in 5m 21s\n- glance-tox-functional-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/3c18f9b0609f43f6b872352d2e3aa719 : SUCCESS in 3m 23s\n- glance-tox-functional-py36-keystone-tips https://zuul.opendev.org/t/openstack/build/d999679e63c04a77b58d9c0efbc7ef65 : SUCCESS in 10m 03s\n- glance-tox-py37-glance_store-tips https://zuul.opendev.org/t/openstack/build/c765f33e862f4dfbbdb8a8e42c447104 : SUCCESS in 3m 56s\n- glance-tox-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/7a3a6eaf6e2b4c90abf9de77d8679a6d : SUCCESS in 5m 14s\n- glance-tox-functional-py37-glance_store-tips https://zuul.opendev.org/t/openstack/build/7e449c0e486b490b8be675fedb4f75e7 : SUCCESS in 3m 16s\n- glance-tox-functional-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/f2f04505dcf04112b0e85e1456b735a3 : SUCCESS in 9m 23s\n- glance-tox-py37-cursive-tips https://zuul.opendev.org/t/openstack/build/b579b42e5e194c78b33755745ce04dc1 : SUCCESS in 4m 01s\n- glance-tox-py36-cursive-tips https://zuul.opendev.org/t/openstack/build/a992164f484b483b831cf40acbc1f9db : SUCCESS in 5m 24s\n- glance-tox-functional-py37-cursive-tips https://zuul.opendev.org/t/openstack/build/a63f486886a54a6aa9641a60937e118e : SUCCESS in 3m 54s\n- glance-tox-functional-py36-cursive-tips https://zuul.opendev.org/t/openstack/build/5589332e1345408581d3ef41763092b1 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2fe7baa62be07960f43f0cdd74c31580ea2ece42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 15:40:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e04bdfbfc38643c29200fcf531d14ac4 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8bc9d3423aa4bb5a943a3b490f2c3e9 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b3de2695ee46b98b6670fe7e2ea541 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1e3f3606a34129a8e2b74be6b23783 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6519d71dc0ad4d56858b9fef04ed37bc : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c184d179d9aa4db7b7f8b9b5e1144980 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa12fb861a94c74a219da52f92d52c1 : SUCCESS in 10m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44c505688a724a3592b57cbc35b13941 : SUCCESS in 8m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a704887b001b43e290e483a807d487c1 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b02c273924a6c9b88b23b5440a6c9d025249cc66","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-11-22 15:46:27.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThere is an earlier attempt by gmann: https://review.opendev.org/#/c/692043/","accounts_in_message":[],"_revision_number":1},{"id":"390ebd7e0be228289e408ba520000e23061ab544","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 22:34:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea630d2f04eb7d64525ea7eff8d22306b9ec06c0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 22:35:55.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nResponse inline.","accounts_in_message":[],"_revision_number":1},{"id":"721593bfd9797ff5a67747461a613b0472199dc8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 22:38:38.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n@Luigi:\n\u003e There is an earlier attempt by gmann: https://review.opendev.org/#/c/692043/\n\nThanks for pointing this out.  I made him a co-author of this patch.","accounts_in_message":[],"_revision_number":2},{"id":"e72b96713cfa911a4b61264b6a8bfa4d3511e709","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-22 22:46:14.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"d98d3a76213c8c83b2327a0d42f781745fb3bfa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 22:46:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9bb19994c615432682d8ed078c408d4e : SUCCESS in 3m 00s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b8ea3c8209f94780b5e45b9d93adecef : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd3a48b5700c41e09607a8f59f215f9f : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b572a3feb2a4765a46e10aaab637176 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aea5164e53c746f893a33ac4c2c825bb : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ee90b971d434c2caf5246e8e7d01496 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe751918dc7e460897f11f5180de9c9a : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c47fc09a7db0457b9d1a56380ae4b137 : SUCCESS in 9m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e28267dd35bd4bc990c202399efa6c3d : SUCCESS in 8m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f66f94bfa51241938a7d7b7d0a87aa00 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"444106cb395de7db7b0a1bb382558d10bfd21485","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 23:02:46.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- glance-tox-functional-py37-oslo-tips https://zuul.opendev.org/t/openstack/build/efe0b530ce0c4f1b933f6d5b077a6af7 : SUCCESS in 4m 16s\n- glance-tox-functional-py36-oslo-tips https://zuul.opendev.org/t/openstack/build/8e3f982ff94546c193fb2a04dd0b6f22 : SUCCESS in 10m 42s\n- glance-tox-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/c3ed345aa9dd4d1d89c0e9018a6b8dd7 : SUCCESS in 4m 03s\n- glance-tox-py36-keystone-tips https://zuul.opendev.org/t/openstack/build/ecd64a9f09f5432ba4b313b829815da6 : SUCCESS in 5m 23s\n- glance-tox-functional-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/bf2612f59dc246c6b7bfb3fc58dd33c0 : SUCCESS in 3m 20s\n- glance-tox-functional-py36-keystone-tips https://zuul.opendev.org/t/openstack/build/55a750b6e96f4d2e92728e4a441df7a1 : POST_FAILURE in 10m 32s\n- glance-tox-py37-glance_store-tips https://zuul.opendev.org/t/openstack/build/f104a3416fca44dd8f0166ead25466b9 : SUCCESS in 2m 47s\n- glance-tox-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/0a15a4f4672942d69053802c8e28a64e : SUCCESS in 4m 43s\n- glance-tox-functional-py37-glance_store-tips https://zuul.opendev.org/t/openstack/build/e0d0e390613a41ba8980d50db220ad29 : SUCCESS in 3m 00s\n- glance-tox-functional-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/3d63d3823aa0404398dce74e3cdc49c8 : SUCCESS in 8m 56s\n- glance-tox-py37-cursive-tips https://zuul.opendev.org/t/openstack/build/0970d551811a4658b57d2fda94032a60 : SUCCESS in 3m 52s\n- glance-tox-py36-cursive-tips https://zuul.opendev.org/t/openstack/build/8c6ea74dd4ff4f0a9978f2b8968440f8 : SUCCESS in 4m 45s\n- glance-tox-functional-py37-cursive-tips https://zuul.opendev.org/t/openstack/build/79572089337e47eba3e236a9b85321bb : SUCCESS in 3m 20s\n- glance-tox-functional-py36-cursive-tips https://zuul.opendev.org/t/openstack/build/b8f1dc91416c40a4beb7113671211416 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":2},{"id":"acceae5d7c49969c2c3b2f1e883a7d2da6930148","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 03:05:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"300ccb18b5b10eca16333987b0e2c6427885f2d7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 03:05:32.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"07cf7dd23f71f378d533b2048e097d70681d97f5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:16:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- glance-tox-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/cb2d7ef3f58c49c4849ca216042127bd : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":3},{"id":"1ab5a5bb524d76b378e598c4b1d15ac58f091bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:18:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d62806061a344999fb265561fcf98a8 : SUCCESS in 2m 40s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/102e561f96d9452db3f6d00851e22bd5 : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dd0982ee09843fc8cc5d689feb7ccda : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078d5d176e684f1db8e6966483e7f2c0 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1e4954a96040eba0afd5f6e0eb30c5 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0278a9b95a82445788ed7af77ed5b5c0 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39ea684447f4ae29d4bc167dab96016 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/974b87034d574de38fdca75cdb8b84fb : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c199b3385e3f4ee89ee49e3c770fa477 : SUCCESS in 7m 43s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/854ebc6091784676a74db37bbec36905 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":3},{"id":"82688a34228c7230e17d35cf18c9b2149bcedfe8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 03:25:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"249a1749f4fa6a8f04db6b99421f7c1db07458bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 03:41:24.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"99e64c64949ff3defdc4d35a2385c3ef8227b328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:41:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/794703619f3c4103958a0d89d5f2db7e : SUCCESS in 3m 01s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2ba586b3829f4ccda6039bee40d6f5c7 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82ee33af8ff1456b9e9a3130524470e6 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915de0aeb9ba4b049d9afcad93bcb3cb : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac6c044a93084260ac373441ab00f594 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc0d3850d9c94ba588943cc4ba83962c : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81470cdcafca440b9e34678d9658bdbe : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/091244c7bb1c47fdbf6cc7332ed3933c : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92a14e1be07c4332815fa06f12ff287f : SUCCESS in 10m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/66ca269116134e2a8de3f7672a88087d : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":4},{"id":"8da7ce5075e21e5feb4d435494e913a8122fd911","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:54:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- glance-tox-functional-py37-oslo-tips https://zuul.opendev.org/t/openstack/build/18582517af694379b7bd292c1207c136 : SUCCESS in 12m 37s\n- glance-tox-functional-py36-oslo-tips https://zuul.opendev.org/t/openstack/build/ac0f2cc5aef54ecca7102447f590df91 : SUCCESS in 11m 11s\n- glance-tox-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/2dd839bc8c184dd78c3ac8e533364aba : SUCCESS in 5m 31s\n- glance-tox-py36-keystone-tips https://zuul.opendev.org/t/openstack/build/768e3e0a38ce475bbb83533920264241 : SUCCESS in 5m 06s\n- glance-tox-functional-py37-keystone-tips https://zuul.opendev.org/t/openstack/build/8930a1b86d3d4930a84bb44ac413601c : SUCCESS in 8m 57s\n- glance-tox-functional-py36-keystone-tips https://zuul.opendev.org/t/openstack/build/6625af4153e54d6db8722ea45d887103 : SUCCESS in 8m 49s\n- glance-tox-py37-glance_store-tips https://zuul.opendev.org/t/openstack/build/23fafde2af0f4318b7b02618746df986 : SUCCESS in 4m 51s\n- glance-tox-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/844354b107db40a98b5d70579827c710 : SUCCESS in 4m 54s\n- glance-tox-functional-py37-glance_store-tips https://zuul.opendev.org/t/openstack/build/64a857f1845d4ec581ae6c71b42f8871 : SUCCESS in 8m 28s\n- glance-tox-functional-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/780854fd27ad4006abce01a5938084b4 : SUCCESS in 8m 49s\n- glance-tox-py37-cursive-tips https://zuul.opendev.org/t/openstack/build/cc91615c87da438e83c91ce35680d504 : SUCCESS in 4m 50s\n- glance-tox-py36-cursive-tips https://zuul.opendev.org/t/openstack/build/f17fd7d111b944319c3776f8c6316bf9 : SUCCESS in 4m 57s\n- glance-tox-functional-py37-cursive-tips https://zuul.opendev.org/t/openstack/build/4b6492490340475b8c0d03df2598eb45 : SUCCESS in 10m 50s\n- glance-tox-functional-py36-cursive-tips https://zuul.opendev.org/t/openstack/build/9c9ea0095ba849d2900d8b3f9cea9303 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":4},{"id":"185aa4e0a6c91df02efc3ef2c0981f394643765a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 04:07:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a46b5b68e3c340c7cc6a5f70062129a64df612c5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 04:08:09.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"6d602c818e7a07c44d9aa532233fd4fa51b804a7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 04:09:13.000000000","message":"Patch Set 6:\n\nThe experimental jobs logs from PS4 show that the tips jobs are running with the correct python interpreter.","accounts_in_message":[],"_revision_number":6},{"id":"851a6da70403c358a678ebf09edc96b93f5789ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 04:19:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a2a198990464b4fb5c467db95481591 : SUCCESS in 2m 33s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0e6060accd0c40bb988407c93c55ab37 : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca8a1d753aac4a1bbdedc96ba03cd0aa : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f48f81ccf0524476858d1b632ffb1351 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec890f7e73e84b9183f1524e4d9c0daa : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f29b00a05a1e4ca8adf4824a031a17b8 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4cf381c1234260bef72a128a0b3dc5 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e554232ebbdf444b9f4c4b9cdd8e57cb : SUCCESS in 9m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf1cec5e31db454f896b9d8e7cc0bcfc : SUCCESS in 8m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/eaa04146468d4d9c9fa01b91191f6251 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":6},{"id":"62e5434a0e6c10e319cfc79bf4ace4afd23baa8a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 12:25:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"482ebf3f26352d5f239f0fa45d261c62f114b17d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-23 12:56:20.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"12d029c20b817e227aa8be075f61c086239cc80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 13:11:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58e33bf93a3f4303b47551028e8e42fb : SUCCESS in 2m 10s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6becd50f05a4fbc9b9b21976d6add31 : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26ce7056f1114389b8977552bfe5c328 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d05968c427430daf9a925119f694a4 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/695e56a7aff84ee1936c280b7808e38b : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61e445a1d744450e9ffd1648e9e2b297 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2146626d2fc4f98ac57040d40bee705 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbf6ea467f944193b51f93e2a98c57ff : SUCCESS in 9m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9cc72d85a684c3fa7754155e1b29e16 : SUCCESS in 7m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8c352add1f604602a87dcaf0a29641da : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":7},{"id":"09c655971d49eb631fd433d90abfd3c5c784c20b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-04 13:06:03.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\nI have one concern with the Python version classifiers, but I think we are OK to not have any version specific ones while we sort out those concerns.","accounts_in_message":[],"_revision_number":7},{"id":"669fdfa0ae4b2a6ded7ef68e5103f9d6bc1b8119","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 02:15:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bae7a93984663db585f23c9d783fbac87f05ad70","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 02:17:28.000000000","message":"Patch Set 8:\n\n@Sean: the py3 classifiers will be taken care of by https://review.opendev.org/#/c/697969/ and https://review.opendev.org/#/c/697970/","accounts_in_message":[],"_revision_number":8},{"id":"3f530d85857df11abc1253433653d89e6f3471bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 02:39:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e69ad119fe604c569a1c203c44ca93c3 : SUCCESS in 3m 12s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/63b08ebf41cb4734b439e79df94e3a55 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48a854fc2a78452e90c302c210bc1656 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e217a9845ad48968f621196a00bfd4c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be7015a2324d4a1381b0ac63531944a6 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c826591c29af411b9d9ffc40fcf496d8 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c877c3bea40841808ebe4fede2115ab7 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e58b8149772f45c6acb87c011eb1f408 : SUCCESS in 15m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc0e38807c44492d8f9b324d965b37db : SUCCESS in 9m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5984c946ed9e475e8bd0066a7da0a3ab : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":8},{"id":"43ab6fdf2f073ab068ad90a896bc2c792cefa592","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-13 06:23:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8254e0b1eb0d2aaea3b16f9c15a523330cadd1be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-13 06:24:05.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"48d15664857e14a26b4808d4e677119f64fc7e1c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-13 10:51:05.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/07/695707/8/check-not-vote/ext-glance-zuul/1b74892 : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":8},{"id":"d1e2b50a63ecb570943d2f9941c9be68e3534e3f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-17 18:46:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"325fec0f8be5591029426eced96a6c6651a36896","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-17 19:00:08.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/07/695707/9/check-not-vote/ext-glance-zuul/8dab9d8 : FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"728b9fb94050f7b643dfd090096ea02824b545d6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-17 19:09:36.000000000","message":"Patch Set 9: Workflow-1\n\nmade a mistake during the rebase, new PS up shortly","accounts_in_message":[],"_revision_number":9},{"id":"bb6fd91cbd4e7b54b1ca3cf128804f82b4e5261f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:21:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e4173c1aa84b4751a4f8382fb18bb288 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ea51f8dc29a4b37a1695c6eac411b2a : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a734bfe37ba44f3181b5d3f4fafb9e7f : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46de1c1ceefd48228c59676dd6e4591a : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/891d695599b64b9ea475ffedc5407ad4 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a43e6c46e8c448fbba81eace074769c : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b732c8663f6468da28bcd856b57a5d7 : SUCCESS in 10m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bcace4e6dd6540f8a90835b85cc73cfa : SUCCESS in 14m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4eda965dcf42497c930fa5c9863745fd : FAILURE in 14m 53s","accounts_in_message":[],"_revision_number":9},{"id":"f3b306b188d8f001a520d356f92f21dbdc671393","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-17 19:26:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b9dab83b417f8157492e64f23333234db33392e5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-17 19:27:02.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/07/695707/10/check-not-vote/ext-glance-zuul/ea5a195 : FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"20de828c1f4e51694edd2c6bfa6bf2e088036d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:46:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/375f38437f93449c8dd599b85ba9fb18 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79493a6003a140b0a2e291685922e61a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091533cd256f4e96ab1e235930a61472 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f603719ae1c247f390611cfe9a040ef0 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e2ccc1554ba4b2cb035d6e73e0e1655 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e398388c4ea14d2fba2ee22da7b50a00 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/307a6b5387654c6786b3cc64bd7dcd3a : SUCCESS in 9m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e65ea50900904c5a8d9f82a82ae11e60 : SUCCESS in 11m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9446c6018181456c9831c3b5e151c42b : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":10},{"id":"49c2ba3ebcd2689068ea6fefb855e2c64ca5a699","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-17 19:49:45.000000000","message":"Patch Set 10: Code-Review+2\n\nLooks like everything is covered.","accounts_in_message":[],"_revision_number":10},{"id":"fef071f41dca9b264bdca867dd6f7c0e333ddf7a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-17 20:28:17.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f2b7800e7e2fb59af68682753b06d5a06489a6fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-18 05:55:31.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nThank you Brian!!!","accounts_in_message":[],"_revision_number":10},{"id":"da300dac1f9cf49d9514c822257f21fb4ef96aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 05:56:28.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"f0f141143a7f09aecbfae9c7aa3db78e531299e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:37:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"995b254001be2587aba7115b8616fe5641ff40e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:37:32.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/99bf9d3ee5f84dcda03d73b848592da4 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/357f4e4d96ce4a0ca2e10eb6e2da339a : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55b7f89c50c94b00ad1b99a5fba619ab : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e21f7174ed64a2499f3425b50ffb851 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2abb2910a5c4f80b902930700a42ccf : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e8a180ce3c4ac7b663e565414aa3fc : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e831c848f0af43af8c033353e4e94d1a : SUCCESS in 10m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/03e9af5e9e2e483ea17487a1355fa501 : SUCCESS in 8m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/14feb5459d2e431a9f83b70dc301c592 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":10},{"id":"8a8f0ed174413935e59c87057a94db23f7b93273","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:42:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0da86cb0bb554ccfaa224f20ffb77400 : SUCCESS in 1m 24s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f51ed08d0240d1a80a20844851b92e : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/561db80c2d794615a74aa1e6600ae2ab : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"40bd312410d6ff23ea5d8b6d6263f40bd456fffd","revisions":{"0e50dc64ceff2ddbc499120d3de8ab90a8aac882":{"kind":"REWORK","_number":1,"created":"2019-11-22 15:15:40.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/1"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"subject":"[WIP] Remove py2 testing, jobs","message":"[WIP] Remove py2 testing, jobs\n\n(The WIP is that the periodic jobs are being run as experimental\nso we can verify that they are working correctly.)\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\n\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e50dc64ceff2ddbc499120d3de8ab90a8aac882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e50dc64ceff2ddbc499120d3de8ab90a8aac882"}]},"branch":"refs/heads/master"},"4029bc6580621d45040390580a8373280b4b9ad4":{"kind":"REWORK","_number":2,"created":"2019-11-22 22:34:09.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/2"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 22:32:19.000000000","tz":-300},"subject":"[WIP] Remove py2 testing, jobs","message":"[WIP] Remove py2 testing, jobs\n\n(The WIP is that the periodic jobs are being run as experimental\nso we can verify that they are working correctly.)\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4029bc6580621d45040390580a8373280b4b9ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4029bc6580621d45040390580a8373280b4b9ad4"}]},"branch":"refs/heads/master"},"ea01dfb549e56f1c49ee7f59ac5ad6299156e4f6":{"kind":"REWORK","_number":3,"created":"2019-11-23 03:05:08.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/3"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-23 03:04:56.000000000","tz":-300},"subject":"[WIP] Remove py2 testing, jobs","message":"[WIP] Remove py2 testing, jobs\n\n(The WIP is that the periodic jobs are being run as experimental\nso we can verify that they are working correctly.)\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea01dfb549e56f1c49ee7f59ac5ad6299156e4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea01dfb549e56f1c49ee7f59ac5ad6299156e4f6"}]},"branch":"refs/heads/master"},"11bb2d6eabfcbd0f5c963cdaf2bdc64dc552f160":{"kind":"REWORK","_number":4,"created":"2019-11-23 03:25:28.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/4"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-23 03:25:20.000000000","tz":-300},"subject":"[WIP] Remove py2 testing, jobs","message":"[WIP] Remove py2 testing, jobs\n\n(The WIP is that the periodic jobs are being run as experimental\nso we can verify that they are working correctly.)\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11bb2d6eabfcbd0f5c963cdaf2bdc64dc552f160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11bb2d6eabfcbd0f5c963cdaf2bdc64dc552f160"}]},"branch":"refs/heads/master"},"ff50fe8c4978800a9eb331a9688ac9fc24239680":{"kind":"REWORK","_number":5,"created":"2019-11-23 04:07:51.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/5"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-23 04:07:24.000000000","tz":-300},"subject":"[WIP] Remove py2 testing, jobs","message":"[WIP] Remove py2 testing, jobs\n\n(The WIP is that the periodic jobs are being run as experimental\nso we can verify that they are working correctly.)\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff50fe8c4978800a9eb331a9688ac9fc24239680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff50fe8c4978800a9eb331a9688ac9fc24239680"}]},"branch":"refs/heads/master"},"ae8ce50e4e9267011e17c0a36b4b8b5e8aad187a":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-11-23 04:08:09.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/6"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-23 04:07:55.000000000","tz":-300},"subject":"Remove py2 testing, jobs","message":"Remove py2 testing, jobs\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae8ce50e4e9267011e17c0a36b4b8b5e8aad187a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae8ce50e4e9267011e17c0a36b4b8b5e8aad187a"}]},"branch":"refs/heads/master"},"3039e71023bca0ce41f7078b158f8f669e89f826":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-11-23 12:56:20.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/7"}}},"commit":{"parents":[{"commit":"d23d12a46257b5c22c5765a85b89da7748155901","subject":"Migrate grenade jobs to py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d23d12a46257b5c22c5765a85b89da7748155901"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-23 12:55:45.000000000","tz":-300},"subject":"Remove py2 testing, jobs","message":"Remove py2 testing, jobs\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nCloses-bug: #1853693\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3039e71023bca0ce41f7078b158f8f669e89f826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3039e71023bca0ce41f7078b158f8f669e89f826"}]},"branch":"refs/heads/master"},"f95423742b183386a47ae00877ad0f5d2b6e8862":{"kind":"REWORK","_number":8,"created":"2019-12-13 02:15:10.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/8"}}},"commit":{"parents":[{"commit":"3011551dfa70e8bd8c63959b93a6765619f0768b","subject":"Drop old neutron-grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3011551dfa70e8bd8c63959b93a6765619f0768b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-13 02:12:04.000000000","tz":-300},"subject":"Remove py2 testing, jobs","message":"Remove py2 testing, jobs\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nCloses-bug: #1853693\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f95423742b183386a47ae00877ad0f5d2b6e8862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f95423742b183386a47ae00877ad0f5d2b6e8862"}]},"branch":"refs/heads/master"},"7a2dcbd23728dae3cb1e740398f5d31c5fda526a":{"kind":"REWORK","_number":9,"created":"2019-12-17 18:46:42.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/9"}}},"commit":{"parents":[{"commit":"0c6dd98d01f9563886a250cceaca6aab3dab173a","subject":"Merge \"Add classifiers of supporting py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c6dd98d01f9563886a250cceaca6aab3dab173a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-17 18:45:59.000000000","tz":-300},"subject":"Remove py2 testing, jobs","message":"Remove py2 testing, jobs\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nCloses-bug: #1853693\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a2dcbd23728dae3cb1e740398f5d31c5fda526a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a2dcbd23728dae3cb1e740398f5d31c5fda526a"}]},"branch":"refs/heads/master"},"40bd312410d6ff23ea5d8b6d6263f40bd456fffd":{"kind":"REWORK","_number":10,"created":"2019-12-17 19:26:42.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/695707/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/695707/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/695707/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/695707/10"}}},"commit":{"parents":[{"commit":"0c6dd98d01f9563886a250cceaca6aab3dab173a","subject":"Merge \"Add classifiers of supporting py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c6dd98d01f9563886a250cceaca6aab3dab173a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-22 15:11:52.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-17 19:24:24.000000000","tz":-300},"subject":"Remove py2 testing, jobs","message":"Remove py2 testing, jobs\n\nChange tox.ini so that all jobs are python3, and add explicit\nunit/functional testenvs for the two ussuri python runtimes\n(3.6 and 3.7).\n\nRemove py2 jobs from .zuul.yaml and redefine the periodic \"tips\"\njobs so that they are testing with the ussuri python runtimes.\nAdded nodeset: ubuntu-bionic to the abstract base jobs so that\npy3.7 is available.\n\nCo-authored-by: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nCloses-bug: #1853693\nChange-Id: I15467c34fc8466fa1724c5401dc879ae77c1922d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40bd312410d6ff23ea5d8b6d6263f40bd456fffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40bd312410d6ff23ea5d8b6d6263f40bd456fffd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
