)]}'
{"id":"openstack%2Fglance~696457","triplet_id":"openstack%2Fglance~master~I13eaab7ab013f44ce18465bdbdbe8052942570ff","project":"openstack/glance","branch":"master","topic":"specs/copy-existing-image","hashtags":[],"change_id":"I13eaab7ab013f44ce18465bdbdbe8052942570ff","subject":"Copy existing image in multiple stores","status":"MERGED","created":"2019-11-28 07:06:34.000000000","updated":"2020-02-13 14:07:19.000000000","submitted":"2020-02-13 14:05:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696457-1581602725563-25e6796c","meta_rev_id":"b2dee00d2eac4d3fe4c6e09cf98959fdc46bb455","_number":696457,"virtual_id_number":696457,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 14:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-12 15:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-12 11:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-13 12:10:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-02-13 14:07:19.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-29 16:26:47.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-12-02 13:47:26.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2020-01-08 17:53:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-01-29 08:04:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-11 03:49:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-12 15:25:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 12:10:50.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-02-13 14:05:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd848778220619750dbf2e436867da331426da8a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-28 07:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9e3c5555e574e2d543c0ea7910374c7e8a09c03","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-11-28 07:07:35.000000000","message":"Patch Set 1: Workflow-1\n\nPoC to give general idea about how copying will work.","accounts_in_message":[],"_revision_number":1},{"id":"80f564affc05abf64d44fc1e59a5f7dbf828d3a1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-28 08:26:47.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/1/check-not-vote/ext-glance-zuul/3e69fa0 : FAILURE in 28m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cd8836a6968817009e8cdf70517c437ee6b50bba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 08:40:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/98324cec03324d7f992b9199c276f53e : SUCCESS in 5m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b34143481c70495182f4bd84a44fbfb3 : SUCCESS in 1h 22m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/265550741d254764bd1cc155e8498520 : SUCCESS in 59m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f8ac21d6fec6467eaeb9821e5366c8bb : SUCCESS in 1h 00m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd26fe7df3df46e589331c22332583db : SUCCESS in 1h 27m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7e31034d17b41398c42093258b82d14 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708bc5ada96d4009a3faf68e58f53664 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fef3b6d2698f4147b70ebbda4f10d0b2 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3771260da76470bb1b1bb3d50381ed4 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a77801a4f3c45559703ad7b5cdabeb0 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a52da62caa48c7bd1e625e62b22866 : SUCCESS in 5m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a14c25360a5c4680be050dd7311d3d5e : SUCCESS in 9m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/46c2a0b5527c4c2b85a40b9a3ac9d024 : SUCCESS in 9m 43s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/0ab916085c4e4401bf420a1ef9c90290 : FAILURE in 5m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cb4b3aebc2c8493e8535a90bb94a46b4 : SUCCESS in 4m 33s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eddaa8c853bc46c580cd6d623cbea6a7 : SUCCESS in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59d87e878eb95a36f5922c49b458386c7ae107c1","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2019-12-01 12:32:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nThe rest is pretty much optimizations and fine tuning, but I definitely thing we should not transition image state anywhere while we are copying the data between stores.\n\nWorst case scenarios this might take hours and there is no way to consume the image meanwhile. Evacuations/migrations would just fail as the image is not \"active\".","accounts_in_message":[],"_revision_number":1},{"id":"6a522c90587edb6a062c0f61537a90f0f349ea45","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-02 07:30:37.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nHi Erno,\n\nThank you for review, I will work on your suggestions.\nAt the moment I will use the base patch of import multiple stores as it is, till it will be modified with your suggestions.","accounts_in_message":[],"_revision_number":1},{"id":"01464812fe96d0a070f9a4ff892156e729e32ea5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 08:13:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94ca00d9e209fb5ab6abc828897ae5ae75bad4e3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-03 08:15:03.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"860b5cd5708e4171babdc79312e1cdd81062a14f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 09:40:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a19456b1702b425f9d53faec934d8ba6 : SUCCESS in 3m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5dc75c53e3bb446c820265cc0d457fb7 : SUCCESS in 1h 25m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e6e7b0c56a0641e58a8cc308393c2502 : SUCCESS in 56m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cfb5e59252f448bcb876c60ca476807a : SUCCESS in 56m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/490e79c3ffb84eb48b5c42192f2e0a4d : SUCCESS in 1h 15m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82d6e0bb550146409e024b2085153288 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d9fdfa324a44a2ac0735a3219b9c65 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32c8273e596d48038583dddba661c4cf : FAILURE in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/830f1335faaf4972a7f460b7870cc061 : FAILURE in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d07cd4f4eda9485db88c781317e13c67 : FAILURE in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652cdf45a136447cbc329dccd46541cc : SUCCESS in 5m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/874f3000a023496aaf6be7d1a93f472e : FAILURE in 11m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3f40d890d58b46c69aad6b92b664d1d2 : FAILURE in 11m 17s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/a62485a77f3744248dcf0f62ba9b625d : FAILURE in 5m 43s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5d9548bfdd27436eaff76a49be5e475f : SUCCESS in 4m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c44f8e7d1a0c42a4a9b936319a0c416e : SUCCESS in 1h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dddabd74d4dfad08b36ae207bf12ae3be15d9f5a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-04 06:10:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"000b9ff299c419b5e7a73fefd1171a2cc7c0759a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-04 06:47:50.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/3/check-not-vote/ext-glance-zuul/cfba03d : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"9dc5f33d18c1dcb3303e51d88bec4e7c0fabc680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 08:49:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c472d659700d4dc7887e9677b71ff8c1 : SUCCESS in 6m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/905bf13a75cc4eb48cd9701a0d569ef1 : TIMED_OUT in 2h 06m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/12835b31ab3f4b868f3f701b37dd8ed8 : SUCCESS in 1h 09m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/99580a1e116040bc8c06a20437d60b0c : SUCCESS in 1h 05m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c97171789724c1096c9a08482948ad8 : SUCCESS in 1h 21m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33c37615a37d41e1bd1556851bc3cb2d : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dcf8fcf919545cb8ef5241c2134c81b : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ad9020f9d1747c983e3525986f6e929 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a97a69239dd2482a8af238cf8a12399f : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54c1f8dde4774ddda4d7be9ce3e1a396 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ea74fd8def4d9d9aec86479c4ece90 : SUCCESS in 10m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/51c7b44efa2c43e69d40692b64eae97c : SUCCESS in 14m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a461935d81ac48478da39bc42296b127 : SUCCESS in 17m 40s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/8a6e2469a58e45beb71010165fae727b : FAILURE in 10m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0360dcf821d1489d9d52261d6da39fff : SUCCESS in 5m 27s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8432a98f80424e31b05d50d7f6d8c122 : SUCCESS in 1h 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cdf51e5001a51702a0f500d665a5d1ad5887e94","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-04 09:25:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1746f2175a27b2fb5573e8a55d003acc2907418a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-04 09:30:14.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/3/check-not-vote/ext-glance-zuul/fb9e53a : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"be16985d440acc8b216b700048a1ed3d5b9e92b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 11:31:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2da44170d654e38a16f36c5b4a6e3f4 : SUCCESS in 2m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/69608f7038b54cd19faa61788f98585c : SUCCESS in 1h 35m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b7721c528a014a46ab0f0bb7c3abf689 : SUCCESS in 59m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a6fa9ba27178454fb19956217c0dab1f : SUCCESS in 1h 55m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6e1c6743627408fbf9ee439bfcab0f6 : SUCCESS in 1h 24m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd8c9d5db857400890965ac92f653508 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/147cd0075d574cb9bc12243540d9af4b : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efd8ce7660f449fda42eed1f29d3a306 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81c15af249d948efb31f8006aa9637f0 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b93b8e824604e85a86ac56d9e50d295 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a941eef685447d8881d52f923e5afb : SUCCESS in 6m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5e32d24525ef4236beafc170ff540a92 : SUCCESS in 9m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b8966a31649f4ce2a7ab54f047c8249d : SUCCESS in 9m 24s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/ad8689cf092a44efb35ff5b08164d4bc : FAILURE in 7m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9db3abaa30384275bbda503cfcb65d55 : SUCCESS in 4m 17s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74197bd445274a25a691ad141fe08b75 : SUCCESS in 1h 03m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ebee84e40b878ddc6dc7d21948723f855918777","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2019-12-05 06:04:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d7ebbb5652a130f050d6ff4f37751be9f180256f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-05 06:37:14.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/4/check-not-vote/ext-glance-zuul/fecb127 : FAILURE in 32m 04s","accounts_in_message":[],"_revision_number":4},{"id":"b440373a4bf5556a15aaa18a83b3a52af70eca82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 07:37:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3cc16dd334574a7a8b7ccd506450bb70 : SUCCESS in 4m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/606563e908124abea29d93250463a657 : SUCCESS in 1h 23m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/028cb03fca7942cf962610a476de7362 : SUCCESS in 59m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a8036e1b8efb4a8a93cc5d9c4a5b9fff : SUCCESS in 58m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bc27f60f3db486ca2f2fee839b5ec86 : SUCCESS in 1h 19m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d10ee590e2b45c0af6eddfa1ceedae3 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff631224b4a4666b3897b8cdb1e98f2 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdd71375384f4b11aa5b751b9373e088 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cd8a1905094c8cae8c77b1d52598da : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccb272e78ec842bebb4346ac733de475 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dac67f299464eda882b7cc6e32ea6a7 : SUCCESS in 7m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d3097f72892746f586f0ca12d4d0c240 : SUCCESS in 11m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4cf6e8701b19401eb6155c9f4130fb96 : SUCCESS in 10m 09s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/994e3e09a0a548d181044685345dc7a2 : FAILURE in 5m 54s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cb1d11bff77c44d4a4aafeecb22ca347 : SUCCESS in 5m 32s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/92bc423a3af74b7fb0fe12f10bf35ef1 : SUCCESS in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"91af0dc637a4c01fc5fd8338f18b4720660c20c7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-06 10:01:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ec8bfd64b017352b67eced80b487fdd316d1eb92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-06 10:12:38.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/5/check-not-vote/ext-glance-zuul/c4d28f4 : FAILURE in 10m 45s","accounts_in_message":[],"_revision_number":5},{"id":"030f15f42725b86f5b29c6fbad2d944fb78e3b53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 11:22:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5251f5f9552b4aa5b0e40275d7242e4a : SUCCESS in 2m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/49843722377a4917a72a1bff6153633b : SUCCESS in 57m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba59c754bee94bac97991b114573563e : SUCCESS in 1h 19m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68d3f5dfb3784f388cffea76a6e59498 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d55e53abf34d48a8a6678762c323b4 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/124172f5fc454bcca515ea0bcc150ef6 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07e72cafc1114717b422c6f7f70be48a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107abd81afb446a8b7bb472d318c0926 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d77ff98c790a424fa565bbf2b0676669 : SUCCESS in 8m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a1ea955f820a4764a0cf44d8f2be617e : SUCCESS in 10m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b8ab1460c8a490da61ec2a239b2ab6a : SUCCESS in 5m 22s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea017c71492a464195f37a5105af9abf : SUCCESS in 1h 03m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ceb5db5b7a77dcb211d4d608382e7b235554e952","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-07 09:23:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4ebe94215afde55bce0230487733268eb4a22efd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-07 09:25:26.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/6/check-not-vote/ext-glance-zuul/4f279ac : FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":6},{"id":"746dd83e142064aecf0380aa623e4530e88c50c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 10:36:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3c4bb5bcfb364367828bf8a048ced062 : SUCCESS in 2m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1d3097b1b13a4698b0923bb0c64443fc : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0faeb76202b4f0f8cfc6cf3b9f8c04b : SUCCESS in 1h 11m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e426d2b3899c413d868aaf31e51d1447 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cfcfb7deef46e3b2ed0c5bed8083fd : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4082b15a98fd499b841d26ea132be9dc : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2e138a197a248d49987c077c5d908ab : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eacb3769638a40a09fed6b3f871f52fe : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/624a4b54d9054f15bde5574a12518c39 : SUCCESS in 8m 53s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/557589aafe4a4076a9712deca7946452 : SUCCESS in 8m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/19138316a69d4a1b8bdf121ef9ffa7f8 : SUCCESS in 4m 33s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b1970aaedaa94832a96fd2d45978bbe9 : SUCCESS in 1h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e006a4f7b97da5e7faecfad5055b57be4296ee4f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-09 10:00:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"632d4cccb613a9dd372c3915fb08500eadd5eddd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-09 10:33:50.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/7/check-not-vote/ext-glance-zuul/f8f8c9f : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":7},{"id":"074ff965ccfab7e22569533b628166a0d452935d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:58:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/261e5e10540742998f61f72cce0019b6 : SUCCESS in 6m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7857e50495ad4516919ff5d16b8307d8 : SUCCESS in 57m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e0286eaa58e463189c8967e44792633 : SUCCESS in 1h 54m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fd401f289e34757b5c899b1a535290a : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35023cf604d546c98f3a58b49994c300 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2604d65060b04099815b905f752501d6 : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7f0231aedc5412bb1c3a3db7154c75c : FAILURE in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c25ffa284c94029858ee281c28e118e : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c69ecd963354743bb9de8ea09b4e557 : SUCCESS in 11m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/23acc47b52cb41dfa07e069625d8f4d7 : SUCCESS in 11m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7906759939d64fc4a386a5548c30a638 : SUCCESS in 4m 40s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/316bf57101254cfcbc9d39420a5cbde9 : SUCCESS in 1h 10m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d1f9bd21eb7443d8dcfee85841c64e4 : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":7},{"id":"1b7d2ec1f3fc8e9286f82f9a1c6ab11de6111a5b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-09 13:37:50.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"9220b42857ef0d7b9dc7ed612d7269044bea4394","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-09 14:03:47.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/7/check-not-vote/ext-glance-zuul/47af37c : FAILURE in 1m 30s","accounts_in_message":[],"_revision_number":7},{"id":"59734be2888f22ddc635758c7b1bdbc644c4bf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 15:03:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e610ae9ccc5046a695dd723dcd3e418d : SUCCESS in 5m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0200355cb0144a73b449d0f86feee723 : SUCCESS in 55m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a366f23b7d34c2ebf02cadf4a6146be : SUCCESS in 1h 17m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2b1f6270a5347018fed01ceca37f093 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071dcb8c241c49ac82c895a3a75cf627 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fae96d3b4399449e97423bd2286eb222 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b09fcf391e614937a122b9ffe1f813fb : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cbcddd1f664aa3b4f7a8a910fb537e : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef07a5ee89f64112a7ef1be2e1b2e18b : SUCCESS in 11m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/caf4e5cf858e4e319bc0b56e7f57a83c : SUCCESS in 11m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4b4185a4bc1a44dd946c7b71b32f1720 : SUCCESS in 5m 48s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a8328e4766c4d468debb3704838ce9c : SUCCESS in 1h 19m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f4ba35a8f5648669ab77752a5391c71 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":7},{"id":"57a8be5a7949b54dbe5d1cdf1ec90d0ed01908b0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-21 10:23:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"40df5820ce0798c26da59447e72d742c5d410084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 14:20:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0f08bbdf874e494eb7bd043b12e229ac : SUCCESS in 6m 39s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8482c5e3ab954501a132c3123c326137 : SUCCESS in 8m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09f4f940e38746dc93829af098e162dd : SUCCESS in 1h 06m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6283ad2b049940eebb4fa72d0e1454fe : SUCCESS in 1h 29m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e2d91425cf44709b6910fd90fa78e11 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89657da168e449448cbce4657519ca74 : FAILURE in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bb2e33fa3944cb2be40a92308c6d0f3 : FAILURE in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c47b05e91dc64f1183768b5c2b777e4a : FAILURE in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a157d093858e48e387b476c777cc41ce : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/963db32e8a2d458ca7838c64e24ac64d : SUCCESS in 10m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f5dfcd58015e4ce4a37b4d3bf9e33a4d : SUCCESS in 10m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dc1eb5a04b9040beb2e859d3c265d4ec : SUCCESS in 6m 45s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a937e7d01fba42bdac9ac4ae92002d2e : SUCCESS in 1h 30m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecf55aa835d149a5b47fabe879dfb34e : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":8},{"id":"e4b2368411ec3a2b91d9a5bf2fba69bb04bc9fe7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-21 15:27:32.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/8/check-not-vote/ext-glance-zuul/66fb6a8 : FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":8},{"id":"825291510a9743983f727d918203108a539b9949","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-21 17:39:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"97c154c5071a0fbd7b8fe30d5e20c30e686698a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:34:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f26784696fa49e5a90e47d792a2cac2 : SUCCESS in 3m 42s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/48ab0cea3deb42eab196ae43d8df95f9 : SUCCESS in 7m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f837c7f822d4e128897fb9735f645d8 : SUCCESS in 1h 14m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ed777962fca4374bdd3804b732aabb5 : SUCCESS in 1h 21m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc4fc3a01c08448a82a91df53c10b7f4 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c411465e3b9045998cb45750e2594b75 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/080d05c8eeed4d88bba6c6fdec0f3fa4 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16c586d539464d2888d30e937c6d71f8 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14d1c03c7c24e3ca9382cc1827a21e9 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e030ef9a1c5d4365b22c08ac8016fac7 : SUCCESS in 10m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e996c258385e4a5ead8c5440b4e769c4 : SUCCESS in 11m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7ca547458ce84ac78a39ee8cced63b3e : SUCCESS in 5m 06s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a3318a628df4e64a8554019ce016126 : FAILURE in 1h 13m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a69ea9df6884960a1d131e1a312ef72 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":9},{"id":"93544bab1fc43ab4593dac7b1e2c2405f1860b92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-21 23:27:15.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/9/check-not-vote/ext-glance-zuul/514f026 : FAILURE in 1m 22s","accounts_in_message":[],"_revision_number":9},{"id":"afdd43c6e952bec4cd4f191a5cc11bb50470d209","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-24 06:11:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0db92603cc7490e8a1a76fea481e47f9cd21b4fd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-24 06:13:14.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/10/check-not-vote/ext-glance-zuul/c21fe36 : FAILURE in 1m 28s","accounts_in_message":[],"_revision_number":10},{"id":"22ee683721736ef47913dba96d3a9e5f9f9b7da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 08:17:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a8b9ec21f98049e283d274e4a63f6b8a : SUCCESS in 5m 02s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6d7d8ad0acf4914bcd2bf7060a610a6 : SUCCESS in 6m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/45f7b6b3d6ba44d7873c50ecbfb3134e : SUCCESS in 55m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6a57e2eb9a1497b84b8da649f2562bd : SUCCESS in 1h 21m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd80c4b83dfa4e528be096d6a068339b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb6e6a0c366400895f8dec5be1ce3d7 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d22a40f905d4cfe93dc84c5245fa349 : SUCCESS in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69229f61ee4d4d95b99d8ad6fd196ac3 : SUCCESS in 13m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e9eb883565464cba73e45c08d5a232 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a893d0ced6e4f68bedb4f9999a9b39c : SUCCESS in 10m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0f6691695f946e9911338d0ff62d382 : SUCCESS in 11m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/07459fef4cc7415aae61a67ada77a478 : SUCCESS in 9m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/997999fb1582481b8b81e706fd454d33 : SUCCESS in 5m 26s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7335e5f075a14113b28baeb1c19724ce : SUCCESS in 1h 22m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0926d9321ad24f70b4a9ce93f7ee5518 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":10},{"id":"890feb6bf00992efe670305f15ac5250aa38f90a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-29 06:18:27.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"63cf12ad534b1e2959ddf5323e6eb2aa07cefd7d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-29 08:04:10.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/696457/11/check-not-vote/ext-glance-zuul/02829c6 : FAILURE in 1m 28s","accounts_in_message":[],"_revision_number":11},{"id":"27036bf0846b2de32f26d0c2bb3a365536649458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 08:12:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db0b90ee2f5142bdb3728fb35d45f3c7 : SUCCESS in 4m 41s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64df85859c9b4526aae88789487a5553 : SUCCESS in 6m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/06d07f54d8424f2699854c9cd84ace4b : SUCCESS in 1h 02m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8cbe8c3cccf4793baf461335fea24c6 : SUCCESS in 1h 17m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76156a39812544e7bbcd33d95c42b9b4 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc2735e3e0f497abe5e5fecfb3ceb18 : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43e99b913bd24cacb96ac3783fdf3067 : SUCCESS in 10m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aab1a28661e34178a750f6a1281f1c53 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/474ed21e699f486aaccd35a09be94e79 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869ee832c46a4f9ab6d825fbc268895b : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca6d18ab8a4742fe9e7ebca91cabba5f : SUCCESS in 11m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70262b6b433e4f019f0a8fda626bea87 : SUCCESS in 11m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c098f3f62f284744bfa332d6043a4e83 : SUCCESS in 15m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a81540a880ef45ea81ea8403edd67c0d : SUCCESS in 8m 25s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f757a87b3a947e89fece9d1cfa8acc6 : SUCCESS in 1h 24m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c729f39b8454daea6c8754fe8a145f2 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":11},{"id":"f638c37d6bd44f5a813087ec29f2378099faf9a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-05 05:30:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"db044d7de3591f86d3cfa71c5561497d246e1a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 07:03:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fce55ec7c61f4e5da81203a5ff028910 : SUCCESS in 3m 17s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/65f1514c4d9149bfb8b3f29eecac5c0a : SUCCESS in 4m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c75c79a3ca45405687f22728975bbf25 : SUCCESS in 1h 14m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c4fe7735de14528b3549ca83bed8a12 : SUCCESS in 1h 32m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77291badb1f14efe92be2d3f0e0dae77 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d34961a7d64e73bb87476d615830f7 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3700d31172424d6eb309545a193052b3 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/433b92ff43544fa0a8d5ec6c51ef26fb : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d5a911728a2436fafd07cefe88ea4b1 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc71af0246f49f9ae7758ee19b64dda : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e35d629df045bea12a4418b10b7551 : SUCCESS in 8m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d548c30f187e4877809f5b732dd0dabd : SUCCESS in 8m 34s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/07c8ba5b8f3446889581913560673dee : SUCCESS in 9m 57s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/02ddfb1d4ac9412baa61e5c2d842e70a : SUCCESS in 4m 46s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/124744958b8646468a7fef23a943987a : SUCCESS in 1h 09m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd42e021d3e844c1929827bd79c436cb : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":12},{"id":"6ad8bbf99c24f9bb615bc13e7d382f96481aa1f4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-05 09:48:26.000000000","message":"Patch Set 12: Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"fc74598dbc38171ea488552c02fdbee58d91d0fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 05:14:06.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"ad66b81e45fa24e8086402ddd03da92c2cbfda32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 06:36:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e92357daedd4b2e885711af5833e566 : SUCCESS in 3m 01s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/83a0f0f0768e4933958227b1082920bb : SUCCESS in 4m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d67eba95aead41a9a13a341ad9feb0ad : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/817876c74da84f16bd768bfefdc51c03 : SUCCESS in 1h 17m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ba1bc11a2364337afc838b2eac29bf3 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1e84e2956c4e01997c6d4593ec22ab : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef1ee90010294011af33661351367787 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dae236bd550640ff910738ccfcc3e36f : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3a158c1ce6e4f6eb0ea0bc483d86b85 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c2ee996aab434498c5cb0889bea426 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/481ac0d24e5f4ee888ded8465934a82f : SUCCESS in 9m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cb27059e2de482ea927cb49a87c0aa0 : TIMED_OUT in 31m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cfa857ea3da540b7a67825eba4c7025c : SUCCESS in 8m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9a893c8c5678449c97e41589d44bd14a : SUCCESS in 4m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b117d7b38e947c7b043a5617112aed1 : SUCCESS in 1h 05m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5033a1cae71d46839b6b45ebd6811aa4 : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":13},{"id":"c43f280f7c0adc762370b491ef8ecf6f16b50e53","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 07:01:59.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"b82ca41d206b3d135b18f0994eedd379154c8bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 08:34:25.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf229ca777f34d5aa6f271d70fc97974 : SUCCESS in 4m 21s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/991749bef09e466d9be9c75afa99f540 : SUCCESS in 5m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5af713861305495cb09a132f42535ccd : SUCCESS in 1h 00m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acf7b8a3fc88457ab61cc705b8ecbbdd : SUCCESS in 1h 19m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7606a4fd5df64262a1c58f2eac92740a : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ea76cf5c864d62821f33d1cc0aaa1b : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1a9670df0704edeafca109bd5e57988 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff9d279557254675aa17f23580e4a6f1 : FAILURE in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a7ef774656543d5951f7ac6cebf8135 : FAILURE in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34cfc7a269d4a68a756fb508cd19c96 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1accf4e1dadd41038f455554719a9d1b : SUCCESS in 11m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a25de6ab0b5a4e9b982a75af97f7e929 : SUCCESS in 13m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9e525f6a0e6b40b1ba9411ec67c29c78 : TIMED_OUT in 31m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ab953ee4229e42fab87b8eaa087c1324 : SUCCESS in 5m 55s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5feeb2bb73fd4cdb9da9e40c6e65a070 : SUCCESS in 1h 28m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf1c49a26fb34a708536b6c8375e40c4 : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":13},{"id":"74d7b336ac5a6633e2b2291fe81a73b698721be7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 08:41:59.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"66a35670b48ebe005c6df3d540f716ee0d097361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:15:08.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77ccb7d53ad844eebe464dde8fbbf94b : SUCCESS in 4m 14s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bf92fa785fc64330ae31f532b313ac7a : SUCCESS in 6m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/853c156175d64c68bfde1d4fcb46f440 : SUCCESS in 59m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84825fff57944960a96578c5a8ea7b2d : SUCCESS in 1h 25m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f8609ebe3ba49f395b2ac5782ce3aea : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa3fa25de124f05bd9e5834816e80c1 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc8f2dd3ca04080bd74ceffb1e3a127 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/629e7066494c45d696460328fc0e5dc2 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9602452aac841e58a3bbf8069c0728a : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425a8ddf77804a018c2b439c5f9fe3c4 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2196536b5b9a4998afaa072b1d6592e3 : SUCCESS in 11m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43f5243053dc42829b154d5cdc261916 : SUCCESS in 10m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e952bc81cee447b4a2388908cf2def0b : TIMED_OUT in 31m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/acc72f73d43f4a7380603b08a690ad7c : SUCCESS in 5m 35s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67957d91a9664d39964d344ce5a512da : SUCCESS in 1h 04m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20ccddb7203948dcb4b97ff45762e805 : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":13},{"id":"f04dba6c97dd8ca9f2e0cc467fa226b9b472f887","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 11:08:56.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5e52a1fef728824caf5698f8a198ec11281e3c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 12:57:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b1706d450494ea788fba67c43debcad : SUCCESS in 4m 59s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3846bb22f6d04627881f1a5c3e8788d3 : SUCCESS in 7m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f2bccc1212284ff49ffb6666d9649cb1 : SUCCESS in 1h 00m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aae3c59da62b47d1adb50aea771cc57a : SUCCESS in 1h 19m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ce653a80fc446f9be4064d5a62b6dea : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b359dec444449f935f5698db574d35 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1616b04cde34f05ba703ecd4901a84e : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54b1a173f6f64d99804c8898c686f9ec : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f3c481e55314a6b917af9b6a771727d : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5bb84cacc8f47ca94614166d40d95c7 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8d5227cbf624f459a0645cb272b62b0 : SUCCESS in 10m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a59d37e10424229b08dbdeaa883bbd6 : SUCCESS in 12m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/606de15e87214677baa0f0cb2a336455 : TIMED_OUT in 32m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/799da7477dfe4c85ae87f7dd12b66659 : SUCCESS in 6m 45s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aea0eb33384b4c84af4a28b46d95345d : SUCCESS in 1h 01m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da690cf4397e4327afdc67531c7b9bf4 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":14},{"id":"8d9c70157daa77c6cf132a4e84d20a9a6d835eb2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 17:48:00.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"9c1f43080ec5c4aa9f1effe34b3d8b504a43529f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:48:18.000000000","message":"Patch Set 15: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias \u0027tempest-irrelevant-files\u0027\n  in \"openstack/glance/.zuul.yaml@master\", line 233, column 31","accounts_in_message":[],"_revision_number":15},{"id":"55be5e2f48e98dd075439bad55313b9a38b64c1a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 19:29:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8e32ea3db0bdbded553db4153011a19aa0673ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 02:36:31.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/faec78ff018249fcb94d0851ed9ffc8b : SUCCESS in 4m 04s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4aa4e3ee19ec452a911652c815b04c9f : SUCCESS in 9m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f108b2f98f04cef950c7885c1ec6922 : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc2aaa7599e3454f8bc26767ae69b09b : SUCCESS in 1h 14m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b06293c7aa84826a5fff727c687e4d2 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6728f8e5bbba44c687a29f8169228571 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d083d97eb2c4c99b3cc9e8f413465a5 : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d64ba45d82345d0aa6b573d7ac9728f : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6ca0317030a4dc5a0035691ab1ceaf2 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ac0237dea64ccc9b540d283591beca : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06a7b14f17454567ad488d9afd1b4a6b : SUCCESS in 13m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8502f554fd94a07898d1df0b793cf0b : SUCCESS in 10m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/89e7fb8611ff48eab309fb6793220c50 : SUCCESS in 10m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7b4f5eb8671d4c7f85c304127a16d229 : SUCCESS in 7m 11s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3af3a94383ad404c839d706c7fb2275d : SUCCESS in 1h 12m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9addedef6e27489eace3164fcdbdde6f : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":16},{"id":"652d67c4e5884dc40de01d2b3bdea342d452a671","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-11 03:45:23.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"945bdc8e5442f00de2cedf1f1bd0dcdb2f08e17e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-11 14:04:25.000000000","message":"Patch Set 16: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"b94d8d126279cb527b565602a48cc389d741a421","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-11 18:06:35.000000000","message":"Patch Set 16:\n\n(7 comments)\n\nJokke,\n\nThank you for review, fixed your comments and added one inline reply.\nKindly review the new patch.","accounts_in_message":[],"_revision_number":16},{"id":"c7a9942a259de87bc084a1a0a6221b5876dfd417","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-11 19:09:57.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"00b3aacfd2e2f36c1d35db2d19cd4cf2e8f3bdc9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-11 19:11:43.000000000","message":"Patch Set 17: Review-Priority+1","accounts_in_message":[],"_revision_number":17},{"id":"288f53518c7cae4304af61a09ca092f5e1aa5a86","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-11 21:42:12.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"86ec044dc5056314f2eeb81768ae8cb14fbbd25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 23:58:07.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8d3c6667b56449c08f2783e01acd96cf : SUCCESS in 6m 31s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44d6a0c13159455694f1bd4a2d247882 : SUCCESS in 6m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e3cf6b15e1144408725424c40c3b094 : FAILURE in 48m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e39bbc6c2454304b766f053d4d1da76 : SUCCESS in 1h 22m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/576913b75d4745c98e38a803664f542a : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8db8c51891d423fa05fa100b784f272 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ce520e65a94a909127185b03c1f4ca : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8c85bba2ac64826a173837d1297680e : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4579cfbd1d4643248328eed6fb101014 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2942a5e0fe9d4c32a443dac8ae0ba060 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7383a36dd644338b15b0764a6e256cc : SUCCESS in 10m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f25feab410ab4bbcb0ea9796976687af : SUCCESS in 10m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/500e4a2fd85f4f91905cbc18fd5215fa : SUCCESS in 12m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/547627367b7142d99239ba15b4560253 : SUCCESS in 5m 51s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c131aa2f462a4f749ae71c2ac0c16898 : SUCCESS in 1h 04m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcbce1a53fd545759479cbdd8b3d7ba9 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":17},{"id":"49a11bb0bbaa2df855598bba71be0bc26c394819","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-12 05:33:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"e96ef59cfa9e3136116c6319672c16262e9732ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 06:59:54.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7098bc88b4ee4af78ecac816395f1ddb : SUCCESS in 4m 12s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/52034f253f08437298fe7dab8a0dd8fb : SUCCESS in 5m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26430872d0654bfcb36b37d67b0b1fbd : SUCCESS in 1h 04m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d3647f3401e4161b4ff4b8b554a8910 : SUCCESS in 1h 21m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e286a8e44baf490a8731da93269cfd61 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1452dfb791543c3acdf2ec0e7c030f3 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/557c07508a4540d784399a63a6d4a60b : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c28cd11cf41460faeaebe81bd19ce5e : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9da343f02914e8493497cdfce65894d : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4752e610142b4928b3a5e86c4d0895ab : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ba206fc54024a9188359a70e918e02f : SUCCESS in 9m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79491c865b5c4eac9c62734d33baeb13 : SUCCESS in 7m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2b3ae228eebb4e49901b4724f589b006 : SUCCESS in 9m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f91c4c7d66ad4d71b597fa2cbcd465e0 : SUCCESS in 6m 23s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c5129f355434a7ebe350ac4f0c5a578 : SUCCESS in 1h 04m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/726e157bc59144299974871acde04fd3 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":18},{"id":"6f79e18d7b5dd862eb4b8404f6a7157001e6a2ff","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-12 11:57:27.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"aa28e77c7543bd1209d7cdb209792749c16a5376","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-12 15:25:48.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)\n\nLooks good to me. I will leave it to someone else to approve though.","accounts_in_message":[],"_revision_number":18},{"id":"e9065f4beaee174405af30172edffc0f75ff50a8","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-13 12:10:50.000000000","message":"Patch Set 18: Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"eb2396d7658aa9343c9037368b4d404cb85342b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:11:13.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"6c92e43fa2e4049b3e17a6cad0e388e7ca781b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:05:25.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e00bc8d2aebe4196861d07529f188fed : SUCCESS in 5m 30s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d832e3f4a6443cf960486a227a46a75 : SUCCESS in 8m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fee0be5975e640878289e70c2180a0f8 : SUCCESS in 1h 25m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/029902ec4edf48a08b40455c42b20bb2 : SUCCESS in 1h 50m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0379fc7d5694401f83d20316972da8d4 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05699045fa4d41c9bde02cdfbf77c93a : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6114512af14b4555bcc5d2ea4657adba : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6e9f50733e343669c0dcff50c73f231 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea312120bfc42e595c5341210121e55 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dca529e8deed47b298ff74807531d421 : SUCCESS in 10m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54376b6cd570466e8536813d81d49ea6 : SUCCESS in 9m 11s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/462faa91e9334bbdb31338e042f92836 : SUCCESS in 9m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/929ee8c395a54c84b2d8647830c1266a : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":18},{"id":"7d5f385cc248b22f78169a19825b5d607adfcd93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:05:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"68c6b4466a326f508d9b5c8a37422511566003e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:07:19.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6e73d1a42a35482babf81798aa2f5a8b : SUCCESS in 1m 03s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9e152502a54b4fa23950f1ee07b208 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3440e80e1ff84414b64abef4bf7a2ae5 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"1754c9e2b085ba0fc37a4369488c92a40268997a","revisions":{"944538eb6aeebd1367715438e16703f44131f4ca":{"kind":"REWORK","_number":1,"created":"2019-11-28 07:06:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/1"}}},"commit":{"parents":[{"commit":"17664ec6d92c5e133c8eed98fb4a80fb350b7d2e","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17664ec6d92c5e133c8eed98fb4a80fb350b7d2e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-28 06:09:25.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-11-28 07:06:20.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/944538eb6aeebd1367715438e16703f44131f4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/944538eb6aeebd1367715438e16703f44131f4ca"}]},"branch":"refs/heads/master"},"fbbb018ae73290fb79f97c110bf4393addf93e0a":{"kind":"REWORK","_number":2,"created":"2019-12-03 08:13:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/2"}}},"commit":{"parents":[{"commit":"518192ab2aeee031364048fc963677e04becc58e","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/518192ab2aeee031364048fc963677e04becc58e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-03 08:13:04.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-03 08:13:04.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbbb018ae73290fb79f97c110bf4393addf93e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbbb018ae73290fb79f97c110bf4393addf93e0a"}]},"branch":"refs/heads/master"},"54bfffd59558a6a857357265e4ebafec80206ec1":{"kind":"REWORK","_number":3,"created":"2019-12-04 06:10:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/3"}}},"commit":{"parents":[{"commit":"b95652cbb2de12df6dd42276eb22e342dfa27d38","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b95652cbb2de12df6dd42276eb22e342dfa27d38"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-04 06:06:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-04 06:06:41.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54bfffd59558a6a857357265e4ebafec80206ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54bfffd59558a6a857357265e4ebafec80206ec1"}]},"branch":"refs/heads/master"},"672414198a770345e92c2b07c194868ad976b073":{"kind":"REWORK","_number":4,"created":"2019-12-05 06:04:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/4"}}},"commit":{"parents":[{"commit":"7a8b9b70229a5151a60a7ce6ce85e772efb6872c","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a8b9b70229a5151a60a7ce6ce85e772efb6872c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-05 06:04:25.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-05 06:04:25.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/672414198a770345e92c2b07c194868ad976b073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/672414198a770345e92c2b07c194868ad976b073"}]},"branch":"refs/heads/master"},"a45b74cc63eda7cc0d9a2fc70d3b63bb81e292be":{"kind":"REWORK","_number":5,"created":"2020-01-06 10:01:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/5"}}},"commit":{"parents":[{"commit":"9e269042ec501c93d8e5fe623b347f67548c013c","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e269042ec501c93d8e5fe623b347f67548c013c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a45b74cc63eda7cc0d9a2fc70d3b63bb81e292be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a45b74cc63eda7cc0d9a2fc70d3b63bb81e292be"}]},"branch":"refs/heads/master"},"c2fae327c1be49fd11535a28c21c2d3ed8cea0c5":{"kind":"REWORK","_number":6,"created":"2020-01-07 09:23:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/6"}}},"commit":{"parents":[{"commit":"9e269042ec501c93d8e5fe623b347f67548c013c","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e269042ec501c93d8e5fe623b347f67548c013c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-07 09:23:13.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c2fae327c1be49fd11535a28c21c2d3ed8cea0c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c2fae327c1be49fd11535a28c21c2d3ed8cea0c5"}]},"branch":"refs/heads/master"},"9b0740feb0bddac9954b0315b1bdebb6e4665eab":{"kind":"REWORK","_number":7,"created":"2020-01-09 10:00:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/7"}}},"commit":{"parents":[{"commit":"9e269042ec501c93d8e5fe623b347f67548c013c","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e269042ec501c93d8e5fe623b347f67548c013c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-09 09:59:57.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nPending:\n  1. Functional tests\n  2. Documentation\n  3. Release note\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0740feb0bddac9954b0315b1bdebb6e4665eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0740feb0bddac9954b0315b1bdebb6e4665eab"}]},"branch":"refs/heads/master"},"e2030d7d954a5b4eeb338e9d18c6ed531bd89ca6":{"kind":"REWORK","_number":8,"created":"2020-01-21 10:23:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/8"}}},"commit":{"parents":[{"commit":"4e6bc6a0a37f60093312aae68de5aaf79a75bf7e","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e6bc6a0a37f60093312aae68de5aaf79a75bf7e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-21 10:22:34.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nPending:\n  1. Release note\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2030d7d954a5b4eeb338e9d18c6ed531bd89ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2030d7d954a5b4eeb338e9d18c6ed531bd89ca6"}]},"branch":"refs/heads/master"},"73d0183b0a4875c12ff5019ec10563bda9294a86":{"kind":"REWORK","_number":9,"created":"2020-01-21 17:39:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/9"}}},"commit":{"parents":[{"commit":"4e6bc6a0a37f60093312aae68de5aaf79a75bf7e","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e6bc6a0a37f60093312aae68de5aaf79a75bf7e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-21 17:39:03.000000000","tz":0},"subject":"PoC - Copy existing image in multiple stores","message":"PoC - Copy existing image in multiple stores\n\nPending:\n  1. Release note\n\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73d0183b0a4875c12ff5019ec10563bda9294a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73d0183b0a4875c12ff5019ec10563bda9294a86"}]},"branch":"refs/heads/master"},"c1a6845d0c525a1de835f2246983c3b7aacf0d86":{"kind":"REWORK","_number":10,"created":"2020-01-24 06:11:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/10"}}},"commit":{"parents":[{"commit":"4e6bc6a0a37f60093312aae68de5aaf79a75bf7e","subject":"WIP: Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e6bc6a0a37f60093312aae68de5aaf79a75bf7e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-24 06:10:44.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1a6845d0c525a1de835f2246983c3b7aacf0d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1a6845d0c525a1de835f2246983c3b7aacf0d86"}]},"branch":"refs/heads/master"},"429dd1bd279d5693b7ca0f25602313e1a563ab10":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-01-29 06:18:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/11"}}},"commit":{"parents":[{"commit":"a6ed4eb233f33a78b0f6b046f3f2cf682ed00843","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6ed4eb233f33a78b0f6b046f3f2cf682ed00843"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-29 05:49:46.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/429dd1bd279d5693b7ca0f25602313e1a563ab10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/429dd1bd279d5693b7ca0f25602313e1a563ab10"}]},"branch":"refs/heads/master"},"7b4f9104ec6cdf26ffc1055eb2ab5d0e21b20ec2":{"kind":"REWORK","_number":12,"created":"2020-02-05 05:30:05.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/12"}}},"commit":{"parents":[{"commit":"96ff4031c04aff8e3507ab19256e03a15f0aa0e9","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96ff4031c04aff8e3507ab19256e03a15f0aa0e9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-05 05:09:13.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b4f9104ec6cdf26ffc1055eb2ab5d0e21b20ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b4f9104ec6cdf26ffc1055eb2ab5d0e21b20ec2"}]},"branch":"refs/heads/master"},"0f14955fd44d26f81a5be2a489568a8923b11c41":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-02-10 05:14:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/13"}}},"commit":{"parents":[{"commit":"8b7610ca1fa4de504c5115b1f8177cdc0d6007ac","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7610ca1fa4de504c5115b1f8177cdc0d6007ac"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-06 10:00:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 04:57:25.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f14955fd44d26f81a5be2a489568a8923b11c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f14955fd44d26f81a5be2a489568a8923b11c41"}]},"branch":"refs/heads/master"},"890c1bfc6fc9b277aa63ad832f65549cde310c80":{"kind":"REWORK","_number":14,"created":"2020-02-10 11:08:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/14"}}},"commit":{"parents":[{"commit":"92492cf50461e214b777c707148886a8e87f340d","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/890c1bfc6fc9b277aa63ad832f65549cde310c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/890c1bfc6fc9b277aa63ad832f65549cde310c80"}]},"branch":"refs/heads/master"},"421947f37c8eae91350ba7cc981b0f7683bfc256":{"kind":"REWORK","_number":15,"created":"2020-02-10 17:48:00.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/15"}}},"commit":{"parents":[{"commit":"92492cf50461e214b777c707148886a8e87f340d","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 17:47:42.000000000","tz":0},"subject":"WIP: Copy existing image in multiple stores","message":"WIP: Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/421947f37c8eae91350ba7cc981b0f7683bfc256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/421947f37c8eae91350ba7cc981b0f7683bfc256"}]},"branch":"refs/heads/master"},"73b0db773ef7c8b9ddd91558875f1379bdae9f1a":{"kind":"REWORK","_number":16,"created":"2020-02-10 19:29:21.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/16"}}},"commit":{"parents":[{"commit":"92492cf50461e214b777c707148886a8e87f340d","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 19:28:22.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73b0db773ef7c8b9ddd91558875f1379bdae9f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73b0db773ef7c8b9ddd91558875f1379bdae9f1a"}]},"branch":"refs/heads/master"},"fd830d86ea5aa6129974507b5269d23f06deab66":{"kind":"REWORK","_number":17,"created":"2020-02-11 19:09:57.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/17"}}},"commit":{"parents":[{"commit":"40ee49b2174418d46f267bb0d029df86508a51b7","subject":"Merge \"Remove registry related functional and unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40ee49b2174418d46f267bb0d029df86508a51b7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-11 19:09:09.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd830d86ea5aa6129974507b5269d23f06deab66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd830d86ea5aa6129974507b5269d23f06deab66"}]},"branch":"refs/heads/master"},"1754c9e2b085ba0fc37a4369488c92a40268997a":{"kind":"REWORK","_number":18,"created":"2020-02-12 05:33:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/57/696457/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/696457/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/696457/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/696457/18"}}},"commit":{"parents":[{"commit":"40ee49b2174418d46f267bb0d029df86508a51b7","subject":"Merge \"Remove registry related functional and unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40ee49b2174418d46f267bb0d029df86508a51b7"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-12 05:32:46.000000000","tz":0},"subject":"Copy existing image in multiple stores","message":"Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nAPIImpact\nImplements: blueprint copy-existing-image\nChange-Id: I13eaab7ab013f44ce18465bdbdbe8052942570ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1754c9e2b085ba0fc37a4369488c92a40268997a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1754c9e2b085ba0fc37a4369488c92a40268997a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
