)]}'
{"id":"openstack%2Fglance~698793","triplet_id":"openstack%2Fglance~master~I3d3ab9b0172521fb42314f2de33ff02985ad4864","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I3d3ab9b0172521fb42314f2de33ff02985ad4864","subject":"Add deprecation message to policy \"default\" rule","status":"MERGED","created":"2019-12-12 19:45:01.000000000","updated":"2020-04-08 11:57:30.000000000","submitted":"2020-04-08 11:49:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698793-1586346578035-75094fdf","meta_rev_id":"04f0aeb21a3d40a2521264beb89d15b43252e960","_number":698793,"virtual_id_number":698793,"owner":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-07 20:13:29.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 11:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-30 05:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2020-03-30 05:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2020-03-30 07:35:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2020-03-30 05:41:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 15:50:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-30 07:35:04.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-04-07 20:13:29.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 11:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e19b01f390fe631e1d8df5ad47b37188d107ebb","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-12 19:45:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ab901cdb5a6f9b9507ae4062ca2f8c9a1097757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:41:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/490eaede99734474ada6fed1bb7b35f1 : FAILURE in 45m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b952d7e3f3964de7acf583e204b4ac3a : SUCCESS in 1h 06m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60231b5b2a144c78bdf48b7ef6f267a9 : FAILURE in 32m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a0a0bd5726545a3905f4084d0181b02 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9056a62b6848cea67ea0a2a7ccc7f4 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a2f5cd0ee6a4bcd9c3fa984a6f3c260 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/292d4fa299a8459c98e934b7e1586cf7 : SUCCESS in 9m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec1f1a0bdb4342a9946aed7ad0213940 : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcf36963907434bb9f1c2d8ca2a0f93 : SUCCESS in 8m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4011035244014b478bbb7e3e09dd2266 : SUCCESS in 12m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d2100f1d2dff47d8a62aba25e9ee6d05 : SUCCESS in 17m 00s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/7ea02ac374cc4daa9283e54ba0d2b93a : FAILURE in 7m 02s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5599982af08f4e04a51049f0e553c62e : SUCCESS in 7m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1225c5f52bb947769f86ad82e6bc1751 : FAILURE in 32m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b2f6c0e3ccfa7819142224ae3f1e1d2de027be1","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-13 02:27:16.000000000","message":"Patch Set 1: Workflow-1\n\nTest fails because it\u0027s using a released version of oslo.policy, not the one from git with the fix.","accounts_in_message":[],"_revision_number":1},{"id":"34dc2f1586510d5bb5a2545255a5d1408f4ca5b2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-13 07:05:36.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/93/698793/1/check-not-vote/ext-glance-zuul/a5b98f2 : FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"431b6cc4052cff65cd5795bbfb71b54a2c712c73","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-16 14:33:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"42ffb1a2ab4b211f601fc2c85bc18f501e306362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 15:30:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/db576f375ab340dbb11e089885149568 : FAILURE in 33m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7928cc5e89194758a411fea8e80c90d8 : FAILURE in 46m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b24c6541b1d14d77aabb886c52aff934 : FAILURE in 33m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38b1e0cd9abc489fad492951d91a00b3 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb88d31fd9d42bcb1256476df23c624 : SUCCESS in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/353864a3d4d74a75806cb9bfc0dc49b6 : SUCCESS in 9m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27679ecfe88474e986fa6c6e2c77605 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d33502ddcf6470d8f3e613fbd31e30e : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d687e58b5aa4c618d5800976d8fa780 : SUCCESS in 8m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/103c8cf2bde1416983a905d3d2d19945 : SUCCESS in 10m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1ee845b96adf407c9cbe3231e5bb9a00 : SUCCESS in 12m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dfc1b2f39bae48febf8ce7e59c578265 : SUCCESS in 8m 17s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84ed8aef66f14586937fccfc71601813 : FAILURE in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c725ada62701430fc27699210c3082d7327e00bb","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-16 16:10:15.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c8fd0aecd97e4365b312c2cacd47e2e43a23bb84","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-16 16:21:53.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/93/698793/2/check-not-vote/ext-glance-zuul/fecd86d : FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"d04d12e93245b7750036b4a80123ef709936f4ac","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-17 03:47:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"82dc54db48e4388be5636fd423c56052216c4777","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-17 03:48:15.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/93/698793/3/check-not-vote/ext-glance-zuul/3976dc6 : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"e03b27505a58c20b7b6c61cff646310c2a1c4669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 04:45:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e0cc4e7164246e0ba25f80806040f8e : SUCCESS in 3m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/161f99d6399042ad97b7f91e55537145 : FAILURE in 26m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b44f989d9b494a1e87015867b268364e : SUCCESS in 56m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c03f44d0ca294e6baf4dfbe06d5a51e6 : FAILURE in 26m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74579414f8304d8eb10d02d29c4d718c : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbfa83719a14437858ad6a9e7aac7e9 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4145616d96c94b1c81c262a60a5bf718 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9105816041c4499b0e2d2207a1f3977 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/209b3584013a48cf86ed00a47b067f31 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018c8564bf5740beabc38354741d6b17 : SUCCESS in 6m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d417feee0894491688942797931f778a : SUCCESS in 10m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4b9702dbeb4d4f7f97d12c24985a8882 : SUCCESS in 10m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a522ff8530224e56ad80d00a4277a48d : SUCCESS in 4m 18s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a89e22ac4afc4d99b919113ea646566e : FAILURE in 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"19643e5e6453d63bccb37fa4446d0fd9e2e848b0","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-17 14:59:55.000000000","message":"Patch Set 3:\n\nLooks like we are still limited to 2.4.0 in tempest due to upper-constraints.","accounts_in_message":[],"_revision_number":3},{"id":"3bb962573110bbd62bee21bd1c6bdbbda124c6c7","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-17 15:26:23.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3fee57bf392629696f63a76461361faa56c860f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 17:15:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/da04c240ff3346fc95e6a04e5434a398 : SUCCESS in 7m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d2d33459b144b319d5d18d3fba90924 : SUCCESS in 1h 46m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3cb73e096e043049984059af99bec3a : SUCCESS in 53m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99e8a4fa53b64459b02b0e6f554bd695 : SUCCESS in 1h 14m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2acbccf7bc204765af04a361469a9300 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecab11aaf3d749fcbdbfdc6e7d9effb3 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/484de2e173774da684c2fae3e8c8c348 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/480beba647994d6084ba2e65a116a569 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5abd85725654928b6d6de15985ef5d6 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d916001bd31420b8dae4273e3726c83 : SUCCESS in 6m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fefc0c77083c4f37bd94f673995b1f3f : SUCCESS in 11m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6e98322f622a48e7abcdba10f86162cb : SUCCESS in 10m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aa1a5761133244f8bb6b5e5985705911 : SUCCESS in 7m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59e3f8a45abb47e7b17384035a28ae86 : SUCCESS in 1h 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd9c107b421b57fea903ffad617c275cf70f530a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-17 18:56:55.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/93/698793/4/check-not-vote/ext-glance-zuul/9c0335f : FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"23b346132e03ad3ac973f0e3afeb5a24a6f61189","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-01-06 18:01:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b36235d54a25eb2700b5d101e6cba3b04df4192f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:44:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf0123283a4244a299b1623197bf3afd : SUCCESS in 5m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6b6bf8cffc0a4beca1fe5bcc82dbe991 : SUCCESS in 1h 18m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/130ad23abce84507b077dccd4d2cd1a2 : SUCCESS in 1h 19m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/917be17d29f242aeb5b9361fc5679721 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b999c27265c840dd9311a1a6f503ea40 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2836a6745924b4292a9cb76cd98fdbd : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f46feb124a1d410890375f5720f4bb71 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/546d2d468578485888848f710d4ecb54 : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79088d47a3ab408a8b1ca9647f06aa52 : SUCCESS in 11m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/802ec6fea08d41b585ff0a7838db676f : SUCCESS in 9m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9d2b2583c2514033b9836e44f7110e48 : SUCCESS in 7m 00s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97a7181bea2b4c29a01dcd96273b3e03 : SUCCESS in 1h 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"56d36bda4fde07bf6f5184c309681e6ff50d5310","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-07 00:51:50.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/93/698793/5/check-not-vote/ext-glance-zuul/1fc19f2 : FAILURE in 10m 47s","accounts_in_message":[],"_revision_number":5},{"id":"cfd0aed2ab2be6b09fe5b0a84b3b18e2dd0f05be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-28 05:22:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6bad7e685d8f8998954c7eb546276e595f31e2ba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-06 08:58:22.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"b60926ce58fe38b53a1defdb6a121287d6406d4b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-05 15:50:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a76571c2da14d943f0144064298322af40e2be9f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-27 18:31:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f6f10718e84899b249904ceaf948f05a2a5f94df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-27 18:31:45.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"6d30972f65be9848c300a257746551d964f2f334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 20:05:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e500227fad74a00be166244aeb6fd16 : SUCCESS in 4m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/96583e2002114d1f8fdab95dad96eee6 : SUCCESS in 1h 08m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d275f92807084dad9bf812f99de36dd9 : SUCCESS in 1h 19m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fb422290d124fdcb07510f72d068378 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26977839500f49a1ac56c8eb91b9377c : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/944e1684b7ca4da3a5bcc6f58894ac62 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64aef21fb2af4bbf8915cb6d14be591e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f19a5fd4a7147d9a2e7ffa505228b1d : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6309e9143104b429d603f539057ad2a : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f8a50d46c2b4f1b8b7aadf26d399f51 : SUCCESS in 9m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ad428994daf74f0396924e1930fd9923 : SUCCESS in 10m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/77f24390be5c4d4cb0884d21a1a714cc : SUCCESS in 4m 48s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ee3883a4dc94975b7fd6d0b5525d994 : SUCCESS in 1h 24m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ef13c5cfd5842a08450fcba0714626a : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":5},{"id":"7a7bc6f714a70ec3e79e0e912fc3eaea22ba8b9d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-30 05:41:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4f43123f98e16f8e6442eb11e17a3591b4296bf0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-30 05:42:09.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"70cc792ad6701af2b10495f3e8de99e94a2aa149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:20:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/44f27ecc0a5c44e9a66810d1e6aea690 : SUCCESS in 2m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6ea7ed654a74a9c83e97ef786be01f6 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/082d81216b6341f68490e1c912c516ab : SUCCESS in 1h 31m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de388aaacd1042afb767902927ddedfe : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7109f31c954dcaa74c8360723c1fc5 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c41890f0f683400e800139fd924c30ae : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1a87fc7048243348da82bc325df8574 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c99f287ae7ce4fd8b2e86853d07ba42e : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a3a86c3a4c46168d79174cd0c408a2 : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5217fd04caf545ce9d38adf5fdd849fe : SUCCESS in 6m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/03653066a7924a63bcaa32bc7791f2b8 : SUCCESS in 10m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9118c33a838a4a94bffbdb833819a9ab : SUCCESS in 6m 20s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b5c8b0fd534e48b7ab7e369dc9218a3e : SUCCESS in 1h 09m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6659809780f2492580e1efc3d723e3c6 : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":6},{"id":"10b565010936ea4ef79d0f2c72d7d5bfa1a0d77a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-30 07:35:04.000000000","message":"Patch Set 6: Workflow+1\n\nrestoring +A, just rebased patch!!","accounts_in_message":[],"_revision_number":6},{"id":"cb571d286cae81459ef6d2336551e00fd5ccd599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:35:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ba07ce533ffed23314b965fc3f006b0a6133af17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 09:20:39.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/93e4ab9db47843608ed29197b371b2d1 : SUCCESS in 3m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/86c38866050e4284b11e274871a3606c : FAILURE in 48m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91820dd3b9b6419c9acfc598da306ff2 : SUCCESS in 1h 17m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31c66055326b4d82a795a7af898e524f : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9a236e7b94406d9aabcd30215b7c90 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35726431f4164a9ba6741280cccfae6c : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be6559d26b744ea097972ae914bed728 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229cb3f9349b43cfbd441b8ce76cf4f2 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13371616c5ae4a689f5281aa600fe760 : TIMED_OUT in 31m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/94ce2e2550144452b786fd5e389f97ef : SUCCESS in 8m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f130896522a4fa89d9572e9f5d87d49 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":6},{"id":"c9d22fed85e1199d48d0e013f2020c544d28ac39","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-30 09:23:03.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"858c90c0dbc7149e450830750a6ef6a1d2a6feb3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-30 09:23:29.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"9cef34a4f9d061aeedc61ca3835e6cb6ea15050f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 12:16:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1da1688221834682ae02bba3fffe4644 : SUCCESS in 3m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a6daff2434674793ad1c9c51f0e23974 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35a7c6e81c8b4bd7a57a8d7be46fbf13 : SUCCESS in 1h 12m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0ddc129fb3c4bcb8af4f54d0432d70f : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d48d6cf75234c98898ed9640c72d92e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65595225f577428caac7b04045aed68b : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34dabb9534cd48489e9dfd1c94bb131a : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/516e39aee5304b41a97ef944e0de1d08 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f570c44a06472e8c844d59f687a2a4 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f30293ead6ad484da58d2bce1c590bf9 : TIMED_OUT in 31m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d275cedde7d04c389ef166675806d602 : SUCCESS in 8m 52s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3969771d97784f9a9411ea98ce0b3e53 : SUCCESS in 4m 11s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/03eac35b83044c01a70043cdeba22824 : SUCCESS in 57m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/921ed890dc7e43e586e166c5093327d7 : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":6},{"id":"d8f03557a67569b53c041c2c9afa5f7b19ce2599","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-31 15:41:07.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c326f21f89ae38f6d42f155224e8662d483d74a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 18:42:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a55e80b7960444f58a17484c544dd16f : SUCCESS in 5m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a8ac80cc102149b7b0001fdeba10cf6d : SUCCESS in 1h 01m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/deabcdd0a11841d793d5f786e1dc1ead : SUCCESS in 1h 17m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b8eef1b07454185aeaa170305f20c1f : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58cc74b382e43248db2eda0b07d70d5 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d48f52e18f1425ead0410454d044cd7 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9201dfe9738c4a12ac740aef5d676c2e : SUCCESS in 12m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e277af9833c4e739991cac99d67c648 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab2ac4645c443048e2cb27a53b136f6 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/912db4bbc1d440baabcbaef51fd259fd : SUCCESS in 9m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1ad985cf79244464a32094fac5c71797 : SUCCESS in 11m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9ad8fde48dae439a8dff6ce4a2683861 : SUCCESS in 6m 25s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d94ef85fa84744d798244117d4f5767d : FAILURE in 56m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d84d8609f4c34daf9b540491d51dbd29 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":6},{"id":"af586032a0a2200d93eb52d3639e759c81994c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 18:43:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7db9c9405ab83d1f6cf4c030b955747f4e48b404","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-31 19:13:58.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"e88e07ef1af334fc15999a43ced8ceacc9a97b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:48:09.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f72f5cde5c144c579b7f032336194c47 : SUCCESS in 4m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/506d0f349db740049c343a7035ad97db : FAILURE in 44m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6bdbf0383d24b2cb3ddb11a7c36bb76 : SUCCESS in 1h 16m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68da7e79daa14565be5f8ab83aa3ff38 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd9ab8422d2423d8d3deb14337345b5 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6929fce6e7ee42aa8a958d9b03f4fffd : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/916542ec55f74d8da1f84a064f5df397 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dd52598ba64891a91208bf9a912f4c : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/119246b5a943402a840eefdc00d9811d : SUCCESS in 8m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/30292074aef947599c3025214452d4b6 : SUCCESS in 9m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dd7b75d6ef647f7b0a4f17a6a4398f6 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":6},{"id":"6f9a78fc8b65b2cd6919651dc438f2a2ea9851cf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 14:55:25.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"66b9683725b84192c398d2904da39219388d3805","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-02 14:55:44.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"05ec31e7aa14146ebc2bb7c4cc8901f1a4a4696b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:05:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cd6a2a6332e4dac913109dd13a51e65 : SUCCESS in 3m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4ec60d8bf868475b9f9590775a3f3b7d : SUCCESS in 1h 06m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a447bed1577c43f69515d1ce55c55d3f : SUCCESS in 1h 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bb9b8b850954754ab0423557d8065b3 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132f8ebd7aa74206b9acba71aeeeb2c9 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c419c6a0ecef423cb5010d7a93eab37c : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a7942f27c154e53b2bf96334a00460f : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f830eaa6c50421d81f7895b48c5efbe : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee777d7144e4244b7cd4a98a21bab8f : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f609dd0d309040edb36588e36b0a9961 : TIMED_OUT in 31m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4fec5b5483c2458199ef195c79a37fae : FAILURE in 12m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c026923479fb4ddca06b0839cd7e197e : SUCCESS in 5m 52s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4c4d86448e654b9db4757fab36713867 : SUCCESS in 1h 05m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fe4759f808a4a6ea33c7376849c81c7 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":6},{"id":"e135663a585e90eecf02865e4582dc4b87b50114","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-03 06:11:05.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"60d6c1d7975aa2f6636f97bffa50dc3f6a0eb402","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-03 06:11:27.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"3afaf056944197158a996998d11615c9d948485c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 08:29:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/59e17eba0840413bb01dfd43dbca54ce : SUCCESS in 4m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3e08e82bf77d432b81f9aed0a8c025d7 : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1c801c50303456aa2538aab051b136d : SUCCESS in 1h 25m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cd8f18492aa419485c5e849b1c5d7fe : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155c79e2a0474138a0f30f3148e9f7aa : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5daf47a4de7f4eda844912aaba928309 : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34c10798e0ad44c9804f6de59cc12f74 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8acc5d4f42bb4c12a215d47b14a5e3fe : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8325f956cbf841cab87065f16b70f9a9 : SUCCESS in 9m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc1e6f2b3c7641668d7730c134d0572f : SUCCESS in 9m 53s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d52c980ef65b4aeea5e98f43b3919067 : TIMED_OUT in 32m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/32e70eff053047c8bb3a517622a1817b : SUCCESS in 7m 28s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6f3a34d730954b15898f905b98e209f7 : SUCCESS in 54m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ac71a3b10f4cb483da81f2ad23bab4 : FAILURE in 1h 01m 42s","accounts_in_message":[],"_revision_number":6},{"id":"74100555503ee0cb911549234fc690cbf7fdb162","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-06 14:27:09.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b2fed80ed9012d8a4d250eeca727742f8af4420a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-06 14:27:31.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"4dfface0d1bef6381e67010fc991c0592434dcb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:00:25.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b2fd8ac7e2d47cdbb2e0ae3ab0a55ba : SUCCESS in 3m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cff9736cf72e419b84420823b88a2b9b : SUCCESS in 1h 14m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3497abbbd39f4f4daaf04ffd7f35c5fd : SUCCESS in 1h 26m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37db536f5ab045d3b0bd561c13823283 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e1e018bc4042bf988aa53adee22c3e : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c05719046a948a2bdcbaf19a50aab0c : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a42762dc6d0c4e7ab86b52c7a710ae98 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acb404b90ba244d79418ca48b1a516f3 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a5329e311944b0a14c009a29c85451 : FAILURE in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4e7ee1bb1e042e28a85e032e2471359 : TIMED_OUT in 31m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b5b2afb09dc54f0a8c666b681865aecc : SUCCESS in 10m 27s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9bdfb4f7c86448dbb62566eb42cc1761 : SUCCESS in 8m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/57ddee9247db41be8d5db2254f51a28a : SUCCESS in 58m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d62b0fe28ddf41b0a0a841d7544e463a : SUCCESS in 57m 59s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file doc/source/admin/apache-httpd.rst\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"5d0a2c0b70bda78a25f506dd563727f781398de8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-06 19:41:11.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6863ff202de9e46c515ac438d0ae3d806ead3e12","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-06 19:41:34.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"dab61b3e83b857e6e70581daa72f9563a196d312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 20:56:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/391c69cadd704425b67853b0afe82e2d : SUCCESS in 3m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b45a9d68ab2d4cedbf5590da13aad36a : SUCCESS in 1h 05m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c5d36ff78c44cf3bfb50df3ba17de00 : SUCCESS in 1h 13m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13c0550d3fb044e8b780eb90ddd6211e : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6e27f2f09c422b86f7d0511f363d86 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9aa2950cdd24d29af0b168367e903cf : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42accdbd8e8c4429a6e09ace0d00ba23 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88e8ccdd43744a7fa500b24d0fb60a26 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f38470d2f042f2a348f0de8e6c7c4b : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb7d8f89fe3d481fa905eb0b41e115fe : TIMED_OUT in 31m 08s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6e9507411d7d4e299fa07005c5e363ff : SUCCESS in 8m 03s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/66832b723c4143ad974fe28441653d1e : SUCCESS in 4m 10s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5e8375b3a0714187b8c75dc1dc79fd41 : SUCCESS in 1h 01m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75254ce6adb34757ad6c6331369785b2 : SUCCESS in 55m 36s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"025f1569653b46d135a957addfb7e8edfee79ea6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-07 05:32:32.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"eb63ec806aacb6e0bd9bfe67cb047f208718b37d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-07 05:33:59.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"786b1cceb0289ae3c698cfbea4f2a0726dcf445b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 07:22:31.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d8aa93ed65d46edbf23f9b86dc7b2e1 : SUCCESS in 4m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a189050a6f7948c0b4f0a906abb814a9 : FAILURE in 1h 06m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1109dc825bd49f6ae29a1428ab8ade1 : FAILURE in 1h 46m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9841716eae674f819ae1c72e413de4c8 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28690fc787474664b1e826344d311369 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55423d4d3024466eb5a59067dc05d8ad : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1568ea7acb8c484099943001209be4ae : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56aa82a02a064a9897405c8f83d1e86d : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66cc4228fc5d49bca0ff7b89db821452 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3efde0c0c7904750a34a868e989bab3f : TIMED_OUT in 32m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f908fe64689a4be4aa6ac90e4cd06009 : SUCCESS in 6m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/89a064c899c24f1587754c9b2e9c15dc : SUCCESS in 4m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/62af86d264f645c68d6ccab50647fed1 : SUCCESS in 1h 13m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0958eb8666514d58b8d0a1e976fae320 : SUCCESS in 1h 02m 32s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"bf9dbfd739ba9099ebd340f45d9106299ab1f92a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-07 07:27:55.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ab03083878b2950fdaed23d99423ff39b6a9bf5e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-07 07:28:14.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"c0a6c934203ddd4dbe69c86e6d5dc80aad51b22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:08:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a43dce15856941938f1ba036a179c819 : SUCCESS in 2m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dcc8b5d0b0f34d42a5546d958c9962e9 : SUCCESS in 1h 12m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d06d111b8d44d88a812cbfc0c5efb15 : SUCCESS in 1h 23m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4659a499dcc14b0ba67cef374874bbe6 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65565ee9af1b48af83bb0b75ef944c63 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51d98cb877af45099286477f995b9155 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1bff537da85439dbe2e71797cd2e240 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/164bdcaee4804e4f96762deaf65fdfdf : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e303326a01dd466595cf664eae60e7e1 : SUCCESS in 9m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c68e17e657d40499b8d884677712b21 : SUCCESS in 7m 58s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bb248cbedef44fcc8f6830a9a7d5ce56 : SUCCESS in 7m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/09902bd547594ff7bb11495f981c5957 : SUCCESS in 6m 30s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4c40b5dbfc5843d88b0bcdc75d56e995 : SUCCESS in 1h 04m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2bbf6443f2745da93f23b1ab9e61015 : SUCCESS in 1h 19m 09s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"67bde6ac990733f46a1c14beb27fc19ee245fca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:09:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"405ea45609c326d394cd69c28317e227dc3417ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 14:38:28.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bfc419b96434d40bea9ca5a1e70b68d : SUCCESS in 7m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/989015a43e3c458083766fe8ab3da66c : SUCCESS in 1h 20m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8618114fedc94416884827a28ba5e74f : FAILURE in 1h 24m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c323cbab247f496e88d691ecf779978d : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f0b924a97648dabab0a0a2722bf822 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b4dd403202341f5bc072a2ce47dbfe6 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/228095ecf91c4a9e9233c363558e30a3 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251f106a7f344b6282cf5f404a122a08 : SUCCESS in 10m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4228c40873394e7fa1aa0619acc89912 : TIMED_OUT in 32m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/871a09a4394948f3b977dbb56a8b6103 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e71b2088478c4185ac7015ba18507faa : SUCCESS in 1h 13m 03s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"8c3e3afeec4a8e10e9eef90f45593d7d28e1655d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-07 19:44:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"cdecb6c6a106763d6f83393f7fdde4dbd93f7e23","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-07 20:13:29.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"a3452a2376e45bb169509dd5ef94f4841c1c1302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:49:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fddd4a69c0334eb98e5c397906724c67 : SUCCESS in 5m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ccbd9bfebe142739dff94a62e1f603d : SUCCESS in 1h 09m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afd1677a16ca4c2fa7e8fe310a1165fa : SUCCESS in 1h 35m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/479fa4c985704928ad30fd0e0b3dfa84 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02cbbe4055b14bd5b1ef740cea13cd79 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdb8e39ad88241678ff152b5b0a2a94f : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d21358eb52b42c99678c403cd2abe24 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfc97d7f095741ff8383d5f070a30b1a : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f2d6cd1a624da69da9d90ec75414c7 : SUCCESS in 10m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea80c9a09c0e41699a90037185ce26ba : SUCCESS in 10m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fe45708ba83d44fcb9752f52f7089ceb : SUCCESS in 14m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/38f59377baf14a15aff5016c45731f64 : SUCCESS in 7m 01s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/47e45a61529d4c87be08b36ba4728e8a : SUCCESS in 1h 12m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51f935a44e1e46589f274b8a2539cf3e : SUCCESS in 1h 07m 12s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"5811267342e00a5720c608a26bccadfe42490940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:02:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f5fc1504b1e3739fac7dd79b61a82b253b6164d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:49:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81533dbdcb7a411883c2032b5d1c89f1 : SUCCESS in 4m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c45586fb2e50423db9c66ea008df2476 : SUCCESS in 1h 11m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f39851c23b2842b2baefbe7dd227cc5c : SUCCESS in 1h 22m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1df8721e3dc4073b08e967122720f10 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be00eeee893b48829c247a798dbc4542 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f12cfb4a697e4340b499db7c7fa7c9cf : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5edf3d0f0ccf4a0a853e687903fdafce : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9ae95275134f75b17524f2c36839f3 : SUCCESS in 9m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38d46b61d02441788d217164c5dccf7c : SUCCESS in 10m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bc649512bff4467cb8afca0d3e9b038e : SUCCESS in 11m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61a4854e71394116a1224f2c32a9b10d : SUCCESS in 1h 04m 39s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":6},{"id":"8ae971cb8b260fe575bb9f7367f51d50ba5ec335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:49:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"04f0aeb21a3d40a2521264beb89d15b43252e960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:57:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5ba33aac0744589a91ea764eb94147 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8acedfd48a3179f71a1b9b67b77eab3223909ade","revisions":{"6dc5a0d2c68d5652b24c02352a07efc03c62196c":{"kind":"REWORK","_number":1,"created":"2019-12-12 19:45:01.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/93/698793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/698793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/698793/1"}}},"commit":{"parents":[{"commit":"029ccf17a0a80cd1272752b901e9f7dbdb19be4c","subject":"Remove empty policy.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/029ccf17a0a80cd1272752b901e9f7dbdb19be4c"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 15:23:35.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 19:34:59.000000000","tz":-300},"subject":"Add deprecation message to policy \"default\" rule","message":"Add deprecation message to policy \"default\" rule\n\nThe default check_str for the \"default\" policy rule has changed. From\nthe Queens release until this series it was \"role:admin\" (prior to\nQueens it was \"@\", which means allow all requests). In order to\naccomodate existing policies as best as possible, the default check_str\nhas been changed back to \"\".\n\nIn the unlikely event that any existing policies are relying on the\ndefault check_str for the \"default\" rule, they need to start explicitly\nspecifying it to maintain similar behaviour. This patch adds a\ndeprecation warning when using the \"default\" rule without overriding the\nvalue in the policy file. This could not be done at the same time as\nchanging the check_str due to a limitation in oslo_config.\n\nChange-Id: I3d3ab9b0172521fb42314f2de33ff02985ad4864\nDepends-On: https://review.opendev.org/698790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dc5a0d2c68d5652b24c02352a07efc03c62196c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dc5a0d2c68d5652b24c02352a07efc03c62196c"}]},"branch":"refs/heads/master"},"cdc9169b6ec2de9d25e8f05c988963473512df29":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-16 14:33:59.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/93/698793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/698793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/698793/2"}}},"commit":{"parents":[{"commit":"89e2276f670f77239ad10842861ad3e6110b5d4a","subject":"Remove empty policy.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89e2276f670f77239ad10842861ad3e6110b5d4a"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 15:23:35.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-16 14:33:43.000000000","tz":-300},"subject":"Add deprecation message to policy \"default\" rule","message":"Add deprecation message to policy \"default\" rule\n\nThe default check_str for the \"default\" policy rule has changed. From\nthe Queens release until this series it was \"role:admin\" (prior to\nQueens it was \"@\", which means allow all requests). In order to\naccomodate existing policies as best as possible, the default check_str\nhas been changed back to \"\".\n\nIn the unlikely event that any existing policies are relying on the\ndefault check_str for the \"default\" rule, they need to start explicitly\nspecifying it to maintain similar behaviour. This patch adds a\ndeprecation warning when using the \"default\" rule without overriding the\nvalue in the policy file. This could not be done at the same time as\nchanging the check_str due to a limitation in oslo_config.\n\nChange-Id: I3d3ab9b0172521fb42314f2de33ff02985ad4864\nDepends-On: https://review.opendev.org/698790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cdc9169b6ec2de9d25e8f05c988963473512df29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cdc9169b6ec2de9d25e8f05c988963473512df29"}]},"branch":"refs/heads/master"},"b8cabdab82a4b4793b2332cadd02d9de30b3032b":{"kind":"REWORK","_number":3,"created":"2019-12-17 03:47:48.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/93/698793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/698793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/698793/3"}}},"commit":{"parents":[{"commit":"89e2276f670f77239ad10842861ad3e6110b5d4a","subject":"Remove empty policy.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89e2276f670f77239ad10842861ad3e6110b5d4a"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 15:23:35.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-17 03:46:42.000000000","tz":-300},"subject":"Add deprecation message to policy \"default\" rule","message":"Add deprecation message to policy \"default\" rule\n\nThe default check_str for the \"default\" policy rule has changed. From\nthe Queens release until this series it was \"role:admin\" (prior to\nQueens it was \"@\", which means allow all requests). In order to\naccomodate existing policies as best as possible, the default check_str\nhas been changed back to \"\".\n\nIn the unlikely event that any existing policies are relying on the\ndefault check_str for the \"default\" rule, they need to start explicitly\nspecifying it to maintain similar behaviour. This patch adds a\ndeprecation warning when using the \"default\" rule without overriding the\nvalue in the policy file. This could not be done at the same time as\nchanging the check_str due to a limitation in oslo.policy.\n\nChange-Id: I3d3ab9b0172521fb42314f2de33ff02985ad4864\nDepends-On: https://review.opendev.org/698790\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8cabdab82a4b4793b2332cadd02d9de30b3032b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8cabdab82a4b4793b2332cadd02d9de30b3032b"}]},"branch":"refs/heads/master"},"045d506acdf64b6ef8102e2a57e7ee8b1f759cdf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-12-17 15:26:23.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/93/698793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/698793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/698793/4"}}},"commit":{"parents":[{"commit":"89e2276f670f77239ad10842861ad3e6110b5d4a","subject":"Remove empty policy.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89e2276f670f77239ad10842861ad3e6110b5d4a"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 15:23:35.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-17 15:25:46.000000000","tz":-300},"subject":"Add deprecation message to policy \"default\" rule","message":"Add deprecation message to policy \"default\" rule\n\nThe default check_str for the \"default\" policy rule has changed. From\nthe Queens release until this series it was \"role:admin\" (prior to\nQueens it was \"@\", which means allow all requests). In order to\naccomodate existing policies as best as possible, the default check_str\nhas been changed back to \"\".\n\nIn the unlikely event that any existing policies are relying on the\ndefault check_str for the \"default\" rule, they need to start explicitly\nspecifying it to maintain similar behaviour. This patch adds a\ndeprecation warning when using the \"default\" rule without overriding the\nvalue in the policy file. This could not be done at the same time as\nchanging the check_str due to a limitation in oslo.policy.\n\nChange-Id: I3d3ab9b0172521fb42314f2de33ff02985ad4864\nDepends-On: https://review.opendev.org/698790\nDepends-On: https://review.opendev.org/699299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/045d506acdf64b6ef8102e2a57e7ee8b1f759cdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/045d506acdf64b6ef8102e2a57e7ee8b1f759cdf"}]},"branch":"refs/heads/master"},"71cd24900985786e53b6d53073343f19694bd2f1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2020-01-06 18:01:52.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/93/698793/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/698793/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/698793/5"}}},"commit":{"parents":[{"commit":"f4520756a1645171e9040f6f3692bd907e667c44","subject":"Remove empty policy.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4520756a1645171e9040f6f3692bd907e667c44"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 15:23:35.000000000","tz":-300},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2020-01-06 17:56:30.000000000","tz":-300},"subject":"Add deprecation message to policy \"default\" rule","message":"Add deprecation message to policy \"default\" rule\n\nThe default check_str for the \"default\" policy rule has changed. From\nthe Queens release until this series it was \"role:admin\" (prior to\nQueens it was \"@\", which means allow all requests). In order to\naccomodate existing policies as best as possible, the default check_str\nhas been changed back to \"\".\n\nIn the unlikely event that any existing policies are relying on the\ndefault check_str for the \"default\" rule, they need to start explicitly\nspecifying it to maintain similar behaviour. This patch adds a\ndeprecation warning when using the \"default\" rule without overriding the\nvalue in the policy file.\n\nThis will result in users who carry over an unsafe policy file from the\nQueens-Train error seeing a warning. Unfortunately it will also result\nin a warning for users who no longer supply a policy file and are\nrelying on the default policies in code.\n\nChange-Id: I3d3ab9b0172521fb42314f2de33ff02985ad4864\nDepends-On: https://review.opendev.org/698790\nDepends-On: https://review.opendev.org/699299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71cd24900985786e53b6d53073343f19694bd2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71cd24900985786e53b6d53073343f19694bd2f1"}]},"branch":"refs/heads/master"},"8acedfd48a3179f71a1b9b67b77eab3223909ade":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-03-30 05:41:49.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/93/698793/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/698793/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/698793/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/698793/6"}}},"commit":{"parents":[{"commit":"2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca","subject":"Merge \"Add decompression import plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-12 15:23:35.000000000","tz":-300},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-03-30 05:41:49.000000000","tz":0},"subject":"Add deprecation message to policy \"default\" rule","message":"Add deprecation message to policy \"default\" rule\n\nThe default check_str for the \"default\" policy rule has changed. From\nthe Queens release until this series it was \"role:admin\" (prior to\nQueens it was \"@\", which means allow all requests). In order to\naccomodate existing policies as best as possible, the default check_str\nhas been changed back to \"\".\n\nIn the unlikely event that any existing policies are relying on the\ndefault check_str for the \"default\" rule, they need to start explicitly\nspecifying it to maintain similar behaviour. This patch adds a\ndeprecation warning when using the \"default\" rule without overriding the\nvalue in the policy file.\n\nThis will result in users who carry over an unsafe policy file from the\nQueens-Train error seeing a warning. Unfortunately it will also result\nin a warning for users who no longer supply a policy file and are\nrelying on the default policies in code.\n\nChange-Id: I3d3ab9b0172521fb42314f2de33ff02985ad4864\nDepends-On: https://review.opendev.org/698790\nDepends-On: https://review.opendev.org/699299\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8acedfd48a3179f71a1b9b67b77eab3223909ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8acedfd48a3179f71a1b9b67b77eab3223909ade"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
