)]}'
{"id":"openstack%2Fglance~700312","triplet_id":"openstack%2Fglance~master~I5f48d06a8fb2afbe02758d7e7963ff6af9b015cc","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I5f48d06a8fb2afbe02758d7e7963ff6af9b015cc","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-12-22 07:57:25.000000000","updated":"2020-01-06 10:43:42.000000000","submitted":"2020-01-06 10:40:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700312-1578307227849-0e1c537b","meta_rev_id":"2eb0651b94901bf798c77ef29c81f58e33f8c4c0","_number":700312,"virtual_id_number":700312,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-04 07:37:08.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 10:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 09:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 09:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-04 07:37:08.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-06 09:54:40.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-01-06 10:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db8c5e9ff68d7707e4eb672687cd3e1f0d45d0f3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-22 07:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f828d5aa7171291881b5e14d4eedb295cb1fb6e4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-22 07:59:06.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/12/700312/1/check-not-vote/ext-glance-zuul/0dfffda : FAILURE in 1m 21s","accounts_in_message":[],"_revision_number":1},{"id":"59bd4257455498c52b049aca7513c904635d00cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 08:08:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d168f8d341d44c269102fd7134ce37f3 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34f284dadf1417d9d51931c359a2cad : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef3574ee9b6542328fc5d67670988f48 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd21ed6e5bc84cd6914016b9c8ba9487 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db7ef45a9da43e78aceb54fe97c6533 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75af885c1258414487694311354ffd84 : SUCCESS in 7m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a75532a052e1431b91cb929dc717a145 : SUCCESS in 10m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/daaa225770724ac7a8b070910e34f0ee : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ece8490ee6b20d92a9006ac160d6b4803dbca78f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-23 07:42:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ce6f8de27f55059d49f38083f2ff7b0d0f9b463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 07:57:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75554cdab9f8401080df08a3c4346932 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c724fb4a6a4f0c8aa9d8a6101d8391 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b24b73bf840549fa870d95b7aa1bd2a5 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a227faf4d1d43bfadaf233c08df16a6 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422c904d21544fa1bfa821b7ff94696b : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/690fdf8caa534abb91b8f492f2380316 : SUCCESS in 9m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c668802631314932ab5ad41c6eb3352d : SUCCESS in 13m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/df5e4144d3724f2196a8cc6d85551baf : SUCCESS in 9m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cc5c4d8105e749f7aace2bebe5987d19 : FAILURE in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8e9c4f1f01b5e81d861450546b699fe9ebf87001","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-23 08:32:51.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/12/700312/2/check-not-vote/ext-glance-zuul/acfb1d2 : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bcbb9fc3bac327a9749387b1c9f33ae124154e09","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-04 07:35:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf948b5c07f2f983c28738f7ba9e4ff1c4cea10b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-04 07:37:08.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/12/700312/3/check-not-vote/ext-glance-zuul/061f28e : FAILURE in 1m 19s","accounts_in_message":[],"_revision_number":3},{"id":"e3617e798dae86324134152c18aab4e15a32cb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 07:51:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e25d841efba44fea20b625d04bc8d0c : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5127555c53674d6091cb3609564a969c : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75fe3b7506a546a6bf02c334963326b8 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/904c0ec87728475e96a98665e0dff84f : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e10116140f9404781e41d4c19e51813 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d804ff0674d4b428ccbc369d6cd0748 : SUCCESS in 9m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eeb21c8078724d0694e8ce826d601d4d : SUCCESS in 9m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c2c93577e97a4634957416cdb9ba9304 : SUCCESS in 11m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4f9e9c9bf9c14196a2bea5b2d1b77cf8 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"f086f9c4067d8f0f828be7e9caa274c1c1a0b2a0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-01-06 09:54:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLooks sane","accounts_in_message":[],"_revision_number":3},{"id":"151b14bbc02d0eda8c978361171fa6789cf2faeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 09:54:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"99ad6ebc877b671582075230116ee4c76568f56c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:40:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37acb4dbd9ca420ebe4103da5bab0d77 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1495239d540d4e98bd0c94f28ff7e1c7 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/866cc8f095364fbfa92d47b56f9854ba : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7eed07a4b32a4016b64e4960713701bb : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a16ef9c6634c138bc1a9632f48121e : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f42e46d9bd2415997948380b6dd522e : SUCCESS in 11m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/236d8dae995142babd950d19300e101a : SUCCESS in 9m 52s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/901c359d7f4044a4948914c06c3392f8 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":3},{"id":"2f9df3b1b5cec2c833ae441e80fdc3750ad92457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:40:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cbfa3636f75cd70c0068cdd915a3d87ade0fe97d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:43:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49db0fa412a64d0fab8c8c7b16033133 : SUCCESS in 2m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cdb900144f244acad814f25031fa7c9 : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2d917f79cfb486225bfac76e85c74f592763930e","revisions":{"d282dab4c2e853cd3a3e55d5c0db50b1c3b6df11":{"kind":"REWORK","_number":1,"created":"2019-12-22 07:57:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/700312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/700312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/700312/1"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 07:57:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-22 07:57:19.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I5f48d06a8fb2afbe02758d7e7963ff6af9b015cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d282dab4c2e853cd3a3e55d5c0db50b1c3b6df11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d282dab4c2e853cd3a3e55d5c0db50b1c3b6df11"}]},"branch":"refs/heads/master"},"774851a345b58b54c1f910ca564c90f6795bfda2":{"kind":"REWORK","_number":2,"created":"2019-12-23 07:42:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/700312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/700312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/700312/2"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-23 07:42:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-23 07:42:23.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I5f48d06a8fb2afbe02758d7e7963ff6af9b015cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/774851a345b58b54c1f910ca564c90f6795bfda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/774851a345b58b54c1f910ca564c90f6795bfda2"}]},"branch":"refs/heads/master"},"2d917f79cfb486225bfac76e85c74f592763930e":{"kind":"REWORK","_number":3,"created":"2020-01-04 07:35:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/700312/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/700312/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/700312/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/700312/3"}}},"commit":{"parents":[{"commit":"d41362517238b9ea3741c4a565e0b73a12c55b92","subject":"Change database migration version to ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d41362517238b9ea3741c4a565e0b73a12c55b92"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-04 07:35:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-04 07:35:27.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I5f48d06a8fb2afbe02758d7e7963ff6af9b015cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d917f79cfb486225bfac76e85c74f592763930e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d917f79cfb486225bfac76e85c74f592763930e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
