)]}'
{"id":"openstack%2Fglance~700997","triplet_id":"openstack%2Fglance~master~Ic80b169d34ae9adc6d1329785f80a0d27d2cb241","project":"openstack/glance","branch":"master","topic":"fix-migration-version","hashtags":[],"change_id":"Ic80b169d34ae9adc6d1329785f80a0d27d2cb241","subject":"Change database migration version to ussuri","status":"MERGED","created":"2020-01-03 08:16:18.000000000","updated":"2020-01-03 18:06:43.000000000","submitted":"2020-01-03 18:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700997-1578074717713-0548d2f5","meta_rev_id":"42b6fbdc42ebd8ea8e8dfef3bc3eebe4c6c81ace","_number":700997,"virtual_id_number":700997,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-03 13:04:11.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-03 18:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-03 14:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-03 16:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-03 16:35:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-01-03 13:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-03 08:17:01.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-01-03 09:10:50.000000000","updated_by":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"reviewer":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"state":"REVIEWER"},{"updated":"2020-01-03 13:04:11.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-03 14:36:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-03 16:35:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-03 18:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df6a2d28bcd4c646e91a4aac381133537bdef428","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-03 08:16:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"428f3f112737b3cde78a61ea40ad22999b5d56b3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-03 08:16:47.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e3bd4549d58787a9240cc07f5d9e4ee4d6a6bf69","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-03 08:17:56.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/97/700997/1/check-not-vote/ext-glance-zuul/05a6357 : FAILURE in 1m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3879000e40aecab5e06c569c4be637acf326a2fa","author":{"_account_id":28367,"name":"Zijian Guo","email":"guozijn@gmail.com","username":"ZijianGuo"},"date":"2020-01-03 09:10:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cc29b6e9e902cd0340a155ee3ee1e9dc09562ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 09:48:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1add91b8e9ce428ca53ebcf12d9cfbb4 : SUCCESS in 52m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9199d610179347239eb2b64e2599a10c : SUCCESS in 1h 12m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56ca1c3ea69245718e66defb7af39fd3 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffde1f54eb844db39679b9e78f06d296 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61c0ee6ea77e4ba895497671ac58d19d : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ef3857491b24be08f4bb1481b1da7f1 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17d3b9496374ae884d1bf2957df21af : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f4d735e95e94dafb058367d694ea846 : FAILURE in 9m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b213d885f2d64afeba5c11521bb781f4 : FAILURE in 10m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/faf2e598f20247b898f9ca35b0976dba : SUCCESS in 4m 18s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db97a8b7631f4d82873c2f278edf195d : SUCCESS in 1h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9d708bb2bf012892b2485b3497e5d940b8e14fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-03 11:42:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03d832b6548f122de583e8f3630375474999a39a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-03 11:44:15.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/97/700997/2/check-not-vote/ext-glance-zuul/01adf7f : FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b73f03190ae3856ad3b6516ad7b884ed3f67b09b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-03 12:08:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19b6e3c84cef962c5a829dd5f81f3b9c334b0d63","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-03 12:10:09.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/97/700997/3/check-not-vote/ext-glance-zuul/d47c133 : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":3},{"id":"713a75ae587701330052025974c61097a2e41e6d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-03 12:24:07.000000000","message":"Patch Set 1:\n\nneed to fix failing functional tests","accounts_in_message":[],"_revision_number":1},{"id":"dcdf21e827588601a3054bf3c3e538a91cf8692f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-03 13:02:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f2b0ef6d9bce2fa786a370feca65d75e95f04da8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-03 13:04:11.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/97/700997/4/check-not-vote/ext-glance-zuul/de70ccd : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":4},{"id":"92f7654341df765976732bee7662cd5ba371a6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 14:20:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/78b3745be0124df09d48ef1def233d54 : SUCCESS in 48m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c98625afe88c40d0bef31ec757209479 : SUCCESS in 1h 16m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc2ff4ea60d44deaa49c7f2c920f8588 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0d30d002a443c093ad69e01b3eca70 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abbd594dfcb7484f991260a0f16a205c : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc3505559a334ea09beb237c480e0d5a : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d90df4d62c45e5816ea6bed555f4ee : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00ac13711b23401dbe7776f362595020 : SUCCESS in 9m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a309d701f3b846269052d91ac13213b6 : SUCCESS in 11m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0bb5dc6e933b41ebb7d611d27c5fb481 : SUCCESS in 4m 21s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40a09b64ca3a4d00af9436426441e944 : SUCCESS in 1h 07m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fc9c9eba593a2e43d517cbbc6d00d8cc10fc33a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-03 14:36:27.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks right to me.","accounts_in_message":[],"_revision_number":4},{"id":"1c0e053c4de5e08933aa769b8ffe4287c9094467","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-03 16:35:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nI don\u0027t think the empty migrations are needed [0], but let\u0027s get the gate unstuck.  We can remove or replace them later.\n\n[0] https://review.opendev.org/#/c/701041/","accounts_in_message":[],"_revision_number":4},{"id":"e7533bb2c4bb917bff97327458db472a316ed585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 16:35:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a8ac440c0f1a650bc42aa1e674099b5349852f86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 18:05:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/10f572efbb9548399f14212b17fcace5 : SUCCESS in 1h 00m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b14c870898d74fbaa513d5c7d7370f49 : SUCCESS in 1h 24m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4c2d9701e5b4997aeb49e5978052478 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abb207c1957474282d653697e268434 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d11a8d798a404058b80339daf12dda55 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b522fb171f74454a5baeccfa1734412 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c282d7051a0f4d8187d9f36cbabadd98 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81706d85b45c41e9a83a463287509bbf : SUCCESS in 9m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e2c671f45bf445a99fcddcfe1529a7ab : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":4},{"id":"d99ea2679ee7d76e8bdeece39710c6bcefc02b63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 18:05:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"78c9a7496f3f59a5b09b8fd5a33e0cf64c47de73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 18:06:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bdb228cfba148b7a9209eec0e8e60fb : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d41362517238b9ea3741c4a565e0b73a12c55b92","revisions":{"a3656b7fd762e0425b636cf4f203d6ab092ea86e":{"kind":"REWORK","_number":1,"created":"2020-01-03 08:16:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/700997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/700997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/700997/1"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 08:15:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 08:15:10.000000000","tz":0},"subject":"Change database migration version to ussuri","message":"Change database migration version to ussuri\n\nChange-Id: Ic80b169d34ae9adc6d1329785f80a0d27d2cb241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3656b7fd762e0425b636cf4f203d6ab092ea86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3656b7fd762e0425b636cf4f203d6ab092ea86e"}]},"branch":"refs/heads/master"},"62cf80cc150f7a308d2e8f8632f23f8ff6745768":{"kind":"REWORK","_number":2,"created":"2020-01-03 11:42:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/700997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/700997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/700997/2"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 08:15:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 11:42:20.000000000","tz":0},"subject":"Change database migration version to ussuri","message":"Change database migration version to ussuri\n\nChange-Id: Ic80b169d34ae9adc6d1329785f80a0d27d2cb241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62cf80cc150f7a308d2e8f8632f23f8ff6745768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62cf80cc150f7a308d2e8f8632f23f8ff6745768"}]},"branch":"refs/heads/master"},"3df56b813e4980373d005aa1543f6bfa96f382dc":{"kind":"REWORK","_number":3,"created":"2020-01-03 12:08:24.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/700997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/700997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/700997/3"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 08:15:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 12:07:35.000000000","tz":0},"subject":"Change database migration version to ussuri","message":"Change database migration version to ussuri\n\nChange-Id: Ic80b169d34ae9adc6d1329785f80a0d27d2cb241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3df56b813e4980373d005aa1543f6bfa96f382dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3df56b813e4980373d005aa1543f6bfa96f382dc"}]},"branch":"refs/heads/master"},"d41362517238b9ea3741c4a565e0b73a12c55b92":{"kind":"REWORK","_number":4,"created":"2020-01-03 13:02:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/700997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/700997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/700997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/700997/4"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 08:15:10.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-03 13:02:14.000000000","tz":0},"subject":"Change database migration version to ussuri","message":"Change database migration version to ussuri\n\nChange-Id: Ic80b169d34ae9adc6d1329785f80a0d27d2cb241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d41362517238b9ea3741c4a565e0b73a12c55b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d41362517238b9ea3741c4a565e0b73a12c55b92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
