)]}'
{"id":"openstack%2Fglance~701041","triplet_id":"openstack%2Fglance~master~Iaa5566eb8f4216ab7c3e5282fff664e2e7953c63","project":"openstack/glance","branch":"master","topic":"fix-migration-version","hashtags":[],"change_id":"Iaa5566eb8f4216ab7c3e5282fff664e2e7953c63","subject":"DNM: checking whether empty migrations are needed","status":"ABANDONED","created":"2020-01-03 16:12:18.000000000","updated":"2021-02-23 22:44:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3ae2cfb4a6ad6c5dc27ca118300564a7729f9eea","_number":701041,"virtual_id_number":701041,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-30 14:37:59.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-01-03 16:17:49.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-03 16:30:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-03 16:17:49.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-03 16:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 14:37:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"c2697bb8d04c8e74b751d80fe9d0aaf252f88764","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-03 16:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0a61dc7c743d34d3d1b01dd430a94d81052d311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 16:12:33.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  found undefined alias \u0027tempest-irrelevant-files\u0027\n  in \"openstack/glance/.zuul.yaml@master\", line 235, column 31","accounts_in_message":[],"_revision_number":1},{"id":"c95f690a28a0beee370755d6f4bbdd783e259be6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-03 16:14:09.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/41/701041/1/check-not-vote/ext-glance-zuul/ade3bcb : FAILURE in 1m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4b1fd549c5fba4587d9db455b82c46c5d2ffa697","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-03 16:15:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f195f8504d8b00f03b3bb91f2285a1069b3734c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-03 16:17:49.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/41/701041/2/check-not-vote/ext-glance-zuul/2e366f9 : FAILURE in 1m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e082654fb61691e2aa1557e25bdbf2ac217ed0f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 16:30:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278392e6aba8449aac50412776357a01 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8a26045ea7540fc805334e5fee267b3 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efcb9318fa7747c3971e8ef93964a2cc : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/446ea65c82c345a099f30c0a1f4c8acb : SUCCESS in 9m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5c2dc4cdbd334eae8f9191e417a74ffa : SUCCESS in 9m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7446e787786e41c38b7c3bbab0cdeb1e : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f5137d8af5a3e766ea14c6eb2f1b5435b3a0d865","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-30 14:37:59.000000000","message":"Patch Set 2:\n\nShould this be abandoned now?","accounts_in_message":[],"_revision_number":2},{"id":"3ae2cfb4a6ad6c5dc27ca118300564a7729f9eea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-23 22:44:58.000000000","message":"Abandoned\n\nno longer needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e217b76ad9b615e06354752e328152d5157f890a","revisions":{"b43ab1683ca82d8d28a6e567f020adb623b3a1d4":{"kind":"REWORK","_number":1,"created":"2020-01-03 16:12:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/701041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/701041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/701041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/701041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/701041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/701041/1"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-01-03 08:15:10.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-01-03 16:11:35.000000000","tz":-300},"subject":"DNM: checking whether empty migrations are needed","message":"DNM: checking whether empty migrations are needed\n\nChange-Id: Iaa5566eb8f4216ab7c3e5282fff664e2e7953c63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b43ab1683ca82d8d28a6e567f020adb623b3a1d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b43ab1683ca82d8d28a6e567f020adb623b3a1d4"}]},"branch":"refs/heads/master"},"e217b76ad9b615e06354752e328152d5157f890a":{"kind":"REWORK","_number":2,"created":"2020-01-03 16:15:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/701041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/701041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/701041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/701041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/701041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/701041/2"}}},"commit":{"parents":[{"commit":"0eacf2da183b641447e1fae703bca65facde91b5","subject":"Refresh Glance example configs for Ussuri milestone 1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0eacf2da183b641447e1fae703bca65facde91b5"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-01-03 08:15:10.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-01-03 16:15:17.000000000","tz":-300},"subject":"DNM: checking whether empty migrations are needed","message":"DNM: checking whether empty migrations are needed\n\nChange-Id: Iaa5566eb8f4216ab7c3e5282fff664e2e7953c63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e217b76ad9b615e06354752e328152d5157f890a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e217b76ad9b615e06354752e328152d5157f890a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
