)]}'
{"id":"openstack%2Fglance~701491","triplet_id":"openstack%2Fglance~stable%2Ftrain~Ib58accd6514e589dccde57fe063815b1ab1ce496","project":"openstack/glance","branch":"stable/train","topic":"bug/1855417-stable/train","hashtags":[],"change_id":"Ib58accd6514e589dccde57fe063815b1ab1ce496","subject":"Staging area not cleared if image is deleted while importing","status":"MERGED","created":"2020-01-08 04:45:41.000000000","updated":"2020-01-09 19:33:43.000000000","submitted":"2020-01-09 19:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701491-1578598308067-7a9322a4","meta_rev_id":"f570fc81b0cbb4dcce3c567ca056db69e13caa9d","_number":701491,"virtual_id_number":701491,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 19:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-09 07:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-01-09 14:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-01-09 14:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"date":"2020-01-08 04:46:18.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-09 07:27:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 04:46:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-09 07:27:05.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-01-09 14:39:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-09 19:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6ed4a1d5b85656e56beba2280c6feef3e480a80","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-08 04:45:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8fcdae64b85acfefe83d0a77a02f54bd6e42030c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 06:16:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3e58fab9420a4233ab657e84afecaf33 : SUCCESS in 1h 28m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e510a596c08746e997483a9399f792cb : SUCCESS in 56m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3fc8c245bada442d8bef8269134d86da : SUCCESS in 47m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c756d507cf324596add1736bc8677593 : SUCCESS in 1h 20m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb4254db544b426ebc6edd685043b1c6 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019457dc1c264844b0c7bf0b3763151a : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42070698026b4cc4b3498156a7994bac : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e96a08e9d774fcfbc36737ebf93a7b8 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d4a2b0ca3ea4ef384b9d8c14e066b52 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e922cce53824259a71a468b774661ee : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1465053fb61f4f36a3550caa121dae5b : SUCCESS in 8m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/495a4fb045814affb1061a9237242861 : SUCCESS in 12m 26s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/4b9fe296aa3842ff89784e35d9844c46 : FAILURE in 4m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/850dfd27f2af4fc2acfe364d460d1eaa : SUCCESS in 4m 25s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/806d0894766a47f9af3df0c244cfa85b : SUCCESS in 1h 08m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8e27c45009ca17a418a10ac7760b1914b7aa927","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-01-09 07:27:05.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nThis backport looks good.","accounts_in_message":[],"_revision_number":1},{"id":"555ce4e327921f175a9925404d7dcecf0d580a87","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-09 14:39:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, and certainly a bug that we want to address.","accounts_in_message":[],"_revision_number":1},{"id":"05689dff29e3f67b05e517dcad16edd1f1263646","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 14:40:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6253cd2b1f1583603cac0fd40fa8b6bcc371ee6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 19:31:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e510ecf84284ed4b60307052fd0df95 : SUCCESS in 1h 27m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5054dd7f533440c29a3843ea4173ea6b : SUCCESS in 1h 10m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bfcce3a9ff4540b4b86cd84191a77b91 : SUCCESS in 1h 06m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbc55ed2e9f9471eb09ce4da6d1e2923 : SUCCESS in 1h 27m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/367e897fc5eb4859a06ea559341e1d54 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20dcf60529f9437aa9a00c764d6fcdaa : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e46c6a1381e546d495c49e949a4f865f : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c215117419c34026a8fc4a7078e25264 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1244e9d02ef54cf1a7e362ee58a5df59 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb2a1da5aa684aa2851a2d6fcb02bf5b : SUCCESS in 7m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7bd57aa03a8e4039bc32b0134964987b : SUCCESS in 11m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6aa3c8acf0f045058891fde8629f28c9 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a261f2b2e4d6559b715134eb9c9ed244cfc5d803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 19:31:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e85ee0389f5b00a4f9a9917cbb6890037ac4a30f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 19:33:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db435c297f744688d66efe129fa0899 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c2bc60820783f76c9421d615d3828a88008ca96","revisions":{"8c2bc60820783f76c9421d615d3828a88008ca96":{"kind":"REWORK","_number":1,"created":"2020-01-08 04:45:41.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/91/701491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/701491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/701491/1"}}},"commit":{"parents":[{"commit":"6e3ced8251cd6e273aa73f553a24fc475b219db5","subject":"Switch to opensuse-15 nodeset","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e3ced8251cd6e273aa73f553a24fc475b219db5"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-06 10:35:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-08 04:45:41.000000000","tz":0},"subject":"Staging area not cleared if image is deleted while importing","message":"Staging area not cleared if image is deleted while importing\n\nIf multiple stores configured in glance and Image is deleted while import\noperation is in progress then image data stays in staging area\n(filesystem backend) and there is no other way than clearing it\nmanually.\n\nModified delete method to delete the data from staging area if image is\ndeleted while import operation is in progress.\n\nChange-Id: Ib58accd6514e589dccde57fe063815b1ab1ce496\nCloses-Bug: #1855417\n(cherry picked from commit 5d15f073718328380f2d847e4abfb9307536d810)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c2bc60820783f76c9421d615d3828a88008ca96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c2bc60820783f76c9421d615d3828a88008ca96"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
