)]}'
{"id":"openstack%2Fglance~701891","triplet_id":"openstack%2Fglance~stable%2Fstein~Ib58accd6514e589dccde57fe063815b1ab1ce496","project":"openstack/glance","branch":"stable/stein","topic":"stein-bug/1855417","hashtags":[],"change_id":"Ib58accd6514e589dccde57fe063815b1ab1ce496","subject":"Staging area not cleared if image is deleted while importing","status":"MERGED","created":"2020-01-10 07:33:52.000000000","updated":"2020-01-14 13:54:55.000000000","submitted":"2020-01-14 13:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701891-1579009975041-bbef5f2b","meta_rev_id":"caf0c3e6516636b4fb6c051c06a6446e5a56b0cf","_number":701891,"virtual_id_number":701891,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-01-10 10:02:29.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 13:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-14 12:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-01-10 11:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-14 12:28:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-01-10 07:39:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 07:39:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-10 11:57:22.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-01-14 12:28:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-14 13:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"988aed99bec6c54620bf74d3d7d64f9aa140b83f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-10 07:33:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"396b883de47cb475efa7b901071fa2d540e369fb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-10 07:38:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fe5524988970dd63169b0020633e1617aa0c44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 09:57:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/98a5c9fa756841bcac630659888ba41f : SUCCESS in 1h 22m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/114f05a4da524dc786558dc80a173052 : FAILURE in 1h 04m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0dd8a17f032c46cf82da6055e9deda2b : SUCCESS in 1h 41m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a2b265e87124e638629b2049d8d7217 : SUCCESS in 1h 56m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a10bf951a9384eac87a5beccd442c733 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cc2e2f0180f4c83869217cc58a66eec : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d274e72a2864a5da001c83b2a8dc7d8 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/262b3de7c4ea4a418ecf822d0680132c : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e61f69983a14a2ca24dbeec8957d01b : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7035097b93fd4cbd8d15ebc176d3faf1 : SUCCESS in 6m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d235e02079546c8aca1cb684ef035cd : SUCCESS in 10m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bf0822add6f4fa8b19a72a45cb6295e : SUCCESS in 8m 34s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/0ce96f26949b4d87a77f5e2e0e5ea157 : FAILURE in 5m 15s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1230af1792cf40e0b58ac898d636d378 : SUCCESS in 1h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a89d05d5a9964961bd2a5759ac45d40c3461408e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-10 10:02:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"203cbbe03f638435e135a6b8a172c73521902fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:35:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/657087d12a1d4e738bd17bb7984e826f : SUCCESS in 1h 26m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91c16e83820e47f7ab1238d867c46217 : SUCCESS in 56m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/792e5af3ef1448beba4806c9c241fcf3 : SUCCESS in 55m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33184a17c0644e3cb51e70a45f0c2190 : SUCCESS in 1h 17m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ccee7eda92a4769bcf861f7b7e9c1f7 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdc341e8f2e460599a713bbc601e097 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/726a86f4e8624e86a479c681694d9388 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b226eba26e44d7a29361f773a944b3 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e40837694db46e59c704c880fe6bd17 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce225546fc0e4ea2965edc9a7f0fdc2c : SUCCESS in 6m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/13ec71fdc1494bcb9f8f2b01b6ee0785 : SUCCESS in 10m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2ddbf76850d41a5bc3cd214ed344e1a : SUCCESS in 9m 45s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/e1fab0aaaf0b4abdaee3e9ee7b0c7266 : FAILURE in 4m 37s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c00ffe73c914971a1c8d2bf5c8d8df9 : SUCCESS in 55m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fcdaff099bfdae020d4ae7aec7d60e913c2074c3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-01-10 11:57:22.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"95284501d32e4a117e54b1a1dee95758cd48e79e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 12:28:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMerged to train and master, good to backport.","accounts_in_message":[],"_revision_number":2},{"id":"de8cde0f93fc29aa128c3e09e2946c481c005e8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:28:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82e492f2a71c534db0ec470a4135575cf950fbba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:52:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/976baa430c244738a9392c178d91f84e : SUCCESS in 1h 23m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a00df6e957604ff89daafa245b2f5c46 : SUCCESS in 1h 02m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/47878dfda68a4f6da8a1cee7fcd4678d : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b97a5ddc91d74f048db3b7ec32fc2401 : SUCCESS in 1h 18m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4d33559c4b54edfa89d74b895aa38ef : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a3880d1edff43a396816d9fcdeb9eee : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d8dfaa453b540bcb50a74c2fb0cb33e : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98f882ed6521456093611c3796244468 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47f069a8353d43b49eeb4a6e884e7465 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3540ae1bf97406daa5f0327ac1a1863 : SUCCESS in 6m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f7615e77601e4299932c4cc08c278656 : SUCCESS in 9m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61fb8f65e80b4c8aaf23a89eb53d2f64 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":2},{"id":"311cd06e78281e6ecb9e798071458e9d19df2729","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:52:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"648f73bc11b29572866dc5ef22bf1c622358da65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:54:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3aac6ee4c204c17bb4d4d2cb18053aa : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"149ea050cc58f39eaf9b4660bb8f0271b99d03da","revisions":{"9b738de8a9ba15ab5d1727dc0213bafea4c756fe":{"kind":"REWORK","_number":1,"created":"2020-01-10 07:33:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/91/701891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/701891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/701891/1"}}},"commit":{"parents":[{"commit":"a027cc453a93c4e64ad81d208b9a6fe755fe8c02","subject":"Merge \"Adding member to image fails for multiple stores\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a027cc453a93c4e64ad81d208b9a6fe755fe8c02"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-06 10:35:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-10 07:32:53.000000000","tz":0},"subject":"Staging area not cleared if image is deleted while importing","message":"Staging area not cleared if image is deleted while importing\n\nIf multiple stores configured in glance and Image is deleted while import\noperation is in progress then image data stays in staging area\n(filesystem backend) and there is no other way than clearing it\nmanually.\n\nModified delete method to delete the data from staging area if image is\ndeleted while import operation is in progress.\nConflicts:\n        glance/api/v2/images.py\n        glance/async_/flows/api_image_import.py\n        glance/common/store_utils.py\n\nNOTE: Conflicts occured due to reserved store capability is not\navailable in Stein.\n\nChange-Id: Ib58accd6514e589dccde57fe063815b1ab1ce496\nCloses-Bug: #1855417\n(cherry picked from commit 5d15f073718328380f2d847e4abfb9307536d810)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b738de8a9ba15ab5d1727dc0213bafea4c756fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b738de8a9ba15ab5d1727dc0213bafea4c756fe"}]},"branch":"refs/heads/stable/stein"},"149ea050cc58f39eaf9b4660bb8f0271b99d03da":{"kind":"REWORK","_number":2,"created":"2020-01-10 07:38:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/91/701891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/701891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/701891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/701891/2"}}},"commit":{"parents":[{"commit":"a027cc453a93c4e64ad81d208b9a6fe755fe8c02","subject":"Merge \"Adding member to image fails for multiple stores\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a027cc453a93c4e64ad81d208b9a6fe755fe8c02"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2019-12-06 10:35:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-01-10 07:38:35.000000000","tz":0},"subject":"Staging area not cleared if image is deleted while importing","message":"Staging area not cleared if image is deleted while importing\n\nIf multiple stores configured in glance and Image is deleted while import\noperation is in progress then image data stays in staging area\n(filesystem backend) and there is no other way than clearing it\nmanually.\n\nModified delete method to delete the data from staging area if image is\ndeleted while import operation is in progress.\nConflicts:\n        glance/api/v2/images.py\n        glance/async_/flows/api_image_import.py\n        glance/common/store_utils.py\n\nNOTE: Conflicts occured due to reserved store capability is not\navailable in Stein.\n\nChange-Id: Ib58accd6514e589dccde57fe063815b1ab1ce496\nCloses-Bug: #1855417\n(cherry picked from commit 5d15f073718328380f2d847e4abfb9307536d810)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/149ea050cc58f39eaf9b4660bb8f0271b99d03da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/149ea050cc58f39eaf9b4660bb8f0271b99d03da"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
