)]}'
{"id":"openstack%2Fglance~703657","triplet_id":"openstack%2Fglance~master~Id24ff7da83ffb8a27767f5f74b12878ecf7bb426","project":"openstack/glance","branch":"master","topic":"Document_behaviour_of_hw_rng_model","attention_set":{},"removed_from_attention_set":{"6962":{"account":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"last_update":"2026-06-24 18:37:43.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-06-28 01:00:54.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id24ff7da83ffb8a27767f5f74b12878ecf7bb426","subject":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027","status":"MERGED","created":"2020-01-21 17:24:27.000000000","updated":"2026-06-24 18:38:56.000000000","submitted":"2026-06-24 18:37:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703657-Document_behaviour_of_hw_rng_model","meta_rev_id":"1862c679919b42a83ab3290114857c8ea32d535e","_number":703657,"virtual_id_number":703657,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 18:37:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-24 16:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-24 16:18:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 19:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-21 22:41:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 18:22:01.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2026-06-24 16:18:48.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"2673001a004cb37d9ded6b682de2def4077128ed","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-01-21 17:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee1afadbd227848e08c281af04256ccecf8baade","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-01-21 17:35:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f5998d7b08b10fe22d75ee0ad18c7960e05c3039","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2020-01-21 17:35:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"abfd18faf2ab86923cb1b08df25a6ccfd550bc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:29:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6949578a1c6f4ec29537c85ace966403 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bfce5a9e74041379afe8e4cc67b5c6d : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/171642d0a14f4c578fea78787dcd45fc : FAILURE in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff56c791919e44c4b907d19abbc6791c : FAILURE in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679f56bb82cf4758a188942ef9d42c78 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/481612345af04c2793c4216b69daba44 : SUCCESS in 10m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f4fe89018593431cab9c37d47fec97be : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":3},{"id":"df3ce2a80f094d0431778b4567d91e23551a7efa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-21 22:41:30.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/57/703657/3/check-not-vote/ext-glance-zuul/f473f0a : FAILURE in 1m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a2acb70b27214613d2bcdcb1abbd35914174080c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-19 18:19:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2295c4da4769063cf6f3439f21c7c69858ea79c3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-09-19 18:22:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f9ebf713c3db504541e43aab4fee1c4f2bac4ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 20:05:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd8a266f532c424ea8e2b8f19a811af5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84306764486c47faa167724ff65aa0e0 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/dfea57d43ef24d729e6e1dcfa5340328 : SUCCESS in 59m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01e147b98b5241608681e8006b77e500 : SUCCESS in 1h 32m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e780d50d9dd476295aa17a4985bbd2e : SUCCESS in 1h 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d913cf94e254c89816de2939d69b715 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67d1d083f904cee9ec69a9c5f935254 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3061615ffead48b6955ee68955509f56 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d9e5ef557c4237af2a0ccebeb150d6 : SUCCESS in 8m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/68823cd043704e0d90da3c075f89a799 : SUCCESS in 13m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a207e0692ee4e0baece5abc3e143ba9 : SUCCESS in 11m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ad3e348d769747b59b73eb592b9af651 : SUCCESS in 9m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4e65506e376f43308fc19958b0418fc5 : SUCCESS in 1h 22m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee828cc45bc9480a9fb96f279eb0d1c4 : SUCCESS in 1h 34m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1f28656958cc410fa6102e7617196bf6 : SUCCESS in 1h 16m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9b3606fe4b14275a3dba54a5c3e2a7d : SUCCESS in 1h 38m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3abe0b37e13e4e4caf8ac1adae472c57 : SUCCESS in 52m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7cd7c1e0b12f4421a69c8797855f270a : SUCCESS in 1h 12m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cbceaf706f604521b010e254f6591df7 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":4},{"id":"a09ae80099d9129d3507be4629995224aca6dafc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-26 18:31:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6eea488d4d3511a8630957e019c24bbe0ee44409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-26 18:39:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3e4990ae11ad712a3319812f4e1598a31f9ad515","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 18:53:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16a3765a5b9417e98400df2ca844a5c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dff25b64f08b4b13a9455de5dd78fbf1 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea1b85a22d914b149bcc263ca8334e12 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/788a1e8ade30498688093e19a4317d47 : SUCCESS in 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"10fbda99d6a7a1f855eb4a3de9d9e1021cf73bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 20:46:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4475f49c812448ea3b6746c287663e9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a15a7d86fafb443e85663071fecd086f : SUCCESS in 4m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/9f3395159770437b9048ac2870984cf8 : SUCCESS in 39m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/704da5227c554979b9e4b60d5a8e59d1 : SUCCESS in 1h 57m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/add29cec6a584cc78bea79d67b2c53b8 : SUCCESS in 1h 00m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d8166ac93647189a60a6015f63833e : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f330161b550a4788afbf4528bf827e23 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79a92c5f52a14cf081054a1d5fd5f317 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f8ea07893704050951e0c9ac90a573d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7b510c55b6497688a5fc373e870058 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4b8affa685b049eaaed39fac5ad1b41e : SUCCESS in 12m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e338f5456b2a46cd99ce17432b55a70f : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b37aff2559af46d787db23549e952f89 : SUCCESS in 7m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8bd37038c16848788ac9391798c706f5 : SUCCESS in 9m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dab1e262933b4a92b0712c3841af653a : TIMED_OUT in 2h 03m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3201a404861e4c9382dfc5a5dabb765a : SUCCESS in 2h 03m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d89571acea83495abf25e5762cfd8f4f : SUCCESS in 1h 48m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7f447489e15940e3919be10c01462631 : SUCCESS in 1h 59m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b317c5ac880b4a39aca51b739cdc2169 : FAILURE in 1h 34m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885f5a7d1141453d88727aa28b3cadba : SUCCESS in 1h 03m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1910e409d2664b0fa957b2220384bcfa : FAILURE in 1h 59m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/045123def1d54c56b57d9c1c440a0afc : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":6},{"id":"8eac1ad124e3b7e6c451fb000896b25dcc81ebdc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-28 01:00:54.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b3d04047af2ca4d03a5107a1819a4829f699d384","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 01:13:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/041763d3118444b8ba02a507e9804616\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f33784850f4b4106b95df567faca0c23 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30a7313e5c324725a6bc86b1370df1a2 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a671d8b7d3924d8a80ea0f769a287915 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af5cc7aac3aa6c5c4c5be01b2c69f2f7dbb613ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 02:44:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cccc923553b7427cb66fe90a61effd0c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af03d7eb39434007b99dc46109cdce18 : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6e9773a268f64f28984d1b8449b2b3f9 : SUCCESS in 50m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24ceb51f822942d7b2e6fe52cf5062e6 : SUCCESS in 1h 22m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98c9535885c54a97b74d87f8f1535184 : SUCCESS in 49m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fda11136254abaa92744a419cfc995 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dce19042cec4636a9059aa19dad7d3b : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b3f8967f3144a18c89f5f055d3b0ce : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90f6f8cd2ce8484e889363983c48221b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2570dd4aa8410e9cdeb58ef430a2d7 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5df487f94ab94d279f1609ed59a7c6d0 : SUCCESS in 12m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f0d7a38926ba4c239adac0bc6cff8d01 : SUCCESS in 9m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e29801ef54a4f54b27baaf5a3abcd6b : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/03e34e7d03a34530b801eb1fda9614a8 : SUCCESS in 8m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/748fcb64d0994892ba3c2e917f0a8fb7 : SUCCESS in 1h 26m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4878ba16d26d4b48bb7ba9c0aa3d29f7 : SUCCESS in 1h 39m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b7238ff17a8943368c6027baef5e2a30 : SUCCESS in 1h 09m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/758ebdd7cde74545b16c981606eb8585 : SUCCESS in 1h 37m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1e986d996dcb4eb7b8e9354e2cd6c9b0 : SUCCESS in 1h 23m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffe5667470ea4f2a9190e3287a7ea1bc : SUCCESS in 51m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7ae0bbec3224105aae1e49a5c668c9c : SUCCESS in 1h 42m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ff35cd15f6948e2a6eb9cca9ce2ce45 : SUCCESS in 44m 12s","accounts_in_message":[],"_revision_number":6},{"id":"1e6167c793abbd8c0df605a2fd8fd8ab2e350e0d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-24 16:18:48.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7f74f4b038b60587f0195219e64421c5be1a4982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:19:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8ab8044f5e0fa9bb0739d809367bd8c495f7c59b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:37:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd9c896e29e45c59d57906ed11114ef\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c721e62df6414485853029960cb8ae6b : SUCCESS in 5m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/4384c8188d324e94967dc96bd24d8b72 : SUCCESS in 51m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f006f515dceb41f0815cbeedf17af15b : SUCCESS in 36m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da1afa27ccf64ffd8331ea413d02201a : SUCCESS in 1h 53m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/813f677d41644843ba759b85d025307f : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/250cfaceb896465f9e9092927fb5665b : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e200bb305cdd42a28ce8564a149fbb27 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40d066f0e6204bbd9ea11f931b5e6ea3 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf39eef05ee450fb3fb4b9a323b14a0 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b7377ef80844b409d77a63087a608d7 : SUCCESS in 7m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/76f6e43ba27847d5bfd8d936825f1c58 : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a3c99ed65e464efa95091bc23134bab4 : SUCCESS in 7m 01s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/846c65ef7f3441f79df5dd69211d508f : SUCCESS in 1h 11m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/71a8e7bd684f4ac5b4247c4d43fdb755 : SUCCESS in 2h 12m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7afb8374b23247bbb482513872d383a7 : SUCCESS in 36m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67ade848432c42ea8e5652fbb33d83fa : SUCCESS in 1h 31m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/021930a9f6804a17853c6925f60f6af6 : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":6},{"id":"9a5140db58a7f439c7b563b49daccc0e7bcb519e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:37:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1862c679919b42a83ab3290114857c8ea32d535e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 18:38:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58622e2bdaa41e9848ff82a2383ab55\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d47358d9b0f24f038c05e83484a1c79e : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333d6c7397b043ba9d56ae69cc27588d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"89f3fd694064bb0ba17f14d868d04e733372929b","revisions":{"3fb5b557ddb40744f52d6479f154c69fc1b4bfa0":{"kind":"REWORK","_number":1,"created":"2020-01-21 17:24:27.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/57/703657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/703657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/703657/1"}}},"commit":{"parents":[{"commit":"2291fe673ec2f875bfc4c49e8ee487c074f1b027","subject":"Merge \"Fix string interpolation to delay by logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2291fe673ec2f875bfc4c49e8ee487c074f1b027"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 16:49:25.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 17:24:15.000000000","tz":60},"subject":"docs: hw_rng_model: Document that it has no effect since Ussuri","message":"docs: hw_rng_model: Document that it has no effect since Ussuri\n\nAlthough, ``hw_rng_model`` is no longer useful, as noted in the Nova\nChange-Id: I28e66c9640c38d23b8c0dbd0b05f5260bfcf6d30 (\"libvirt: Add a\ndefault VirtIO-RNG device to guests\"), we may extend it to take further\nvalues to enable other use cases (details in the Nova commit).\n\nHence we should not yet deprecate it.\n\nChange-Id: Id24ff7da83ffb8a27767f5f74b12878ecf7bb426\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fb5b557ddb40744f52d6479f154c69fc1b4bfa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fb5b557ddb40744f52d6479f154c69fc1b4bfa0"}]},"branch":"refs/heads/master"},"6618487da2bec6fc73bf141237bca81b2515c675":{"kind":"REWORK","_number":2,"created":"2020-01-21 17:35:30.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/57/703657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/703657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/703657/2"}}},"commit":{"parents":[{"commit":"2291fe673ec2f875bfc4c49e8ee487c074f1b027","subject":"Merge \"Fix string interpolation to delay by logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2291fe673ec2f875bfc4c49e8ee487c074f1b027"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 16:49:25.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 17:35:10.000000000","tz":60},"subject":"docs: hw_rng_model: Document that it has no effect since Ussuri","message":"docs: hw_rng_model: Document that it has no effect since Ussuri\n\nAlthough, ``hw_rng_model`` is no longer useful, as noted in the Nova\nChange-Id: I28e66c9640c38d23b8c0dbd0b05f5260bfcf6d30 (\"libvirt: Add a\ndefault VirtIO-RNG device to guests\"), we may extend it to take further\nvalues to enable other use cases (details in the Nova commit).\n\nHence we should not yet deprecate it.\n\nChange-Id: Id24ff7da83ffb8a27767f5f74b12878ecf7bb426\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6618487da2bec6fc73bf141237bca81b2515c675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6618487da2bec6fc73bf141237bca81b2515c675"}]},"branch":"refs/heads/master"},"9ac0a189caf71972adedf2daf32ffc5d7b11a7e4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-21 17:35:55.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/57/703657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/703657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/703657/3"}}},"commit":{"parents":[{"commit":"2291fe673ec2f875bfc4c49e8ee487c074f1b027","subject":"Merge \"Fix string interpolation to delay by logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2291fe673ec2f875bfc4c49e8ee487c074f1b027"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 16:49:25.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 17:35:44.000000000","tz":60},"subject":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027","message":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027\n\nAlthough, ``hw_rng_model`` is no longer useful, as noted in the Nova\nChange-Id: I28e66c9640c38d23b8c0dbd0b05f5260bfcf6d30 (\"libvirt: Add a\ndefault VirtIO-RNG device to guests\"), we may extend it to take further\nvalues to enable other use cases (details in the Nova commit).\n\nHence we should not yet deprecate it.\n\nChange-Id: Id24ff7da83ffb8a27767f5f74b12878ecf7bb426\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ac0a189caf71972adedf2daf32ffc5d7b11a7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ac0a189caf71972adedf2daf32ffc5d7b11a7e4"}]},"branch":"refs/heads/master"},"74b6a7c5a59ed6450265442db8c292894b43bb7d":{"kind":"REWORK","_number":4,"created":"2022-09-19 18:19:08.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/57/703657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/703657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/703657/4"}}},"commit":{"parents":[{"commit":"dc11640d2fea4010cd64f3bac861ab8bc3501044","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc11640d2fea4010cd64f3bac861ab8bc3501044"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 16:49:25.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-09-19 18:18:19.000000000","tz":120},"subject":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027","message":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027\n\nAlthough, ``hw_rng_model`` is no longer useful, as noted in the Nova\nChange-Id: I28e66c9640c38d23b8c0dbd0b05f5260bfcf6d30 (\"libvirt: Add a\ndefault VirtIO-RNG device to guests\"), we may extend it to take further\nvalues to enable other use cases (details in the Nova commit).\n\nHence we should not yet deprecate it.\n\nChange-Id: Id24ff7da83ffb8a27767f5f74b12878ecf7bb426\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74b6a7c5a59ed6450265442db8c292894b43bb7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74b6a7c5a59ed6450265442db8c292894b43bb7d"}]},"branch":"refs/heads/master"},"b12f5c97107f480f60b0ab52787f6b832b65b2b0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-06-26 18:31:03.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/57/703657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/703657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/703657/5"}}},"commit":{"parents":[{"commit":"d47803613d501b18f9fa4f6aa3c899bd16f9ddde","subject":"Handle case for multiple cirros images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d47803613d501b18f9fa4f6aa3c899bd16f9ddde"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 16:49:25.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-26 18:31:03.000000000","tz":0},"subject":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027","message":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027\n\nAlthough, ``hw_rng_model`` is no longer useful, as noted in the Nova\nChange-Id: I28e66c9640c38d23b8c0dbd0b05f5260bfcf6d30 (\"libvirt: Add a\ndefault VirtIO-RNG device to guests\"), we may extend it to take further\nvalues to enable other use cases (details in the Nova commit).\n\nHence we should not yet deprecate it.\n\nChange-Id: Id24ff7da83ffb8a27767f5f74b12878ecf7bb426\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b12f5c97107f480f60b0ab52787f6b832b65b2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b12f5c97107f480f60b0ab52787f6b832b65b2b0"}]},"branch":"refs/heads/master","description":"Rebase"},"89f3fd694064bb0ba17f14d868d04e733372929b":{"kind":"REWORK","_number":6,"created":"2023-06-26 18:39:35.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/57/703657/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/703657/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/703657/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/703657/6"}}},"commit":{"parents":[{"commit":"d47803613d501b18f9fa4f6aa3c899bd16f9ddde","subject":"Handle case for multiple cirros images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d47803613d501b18f9fa4f6aa3c899bd16f9ddde"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2020-01-21 16:49:25.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-26 18:39:28.000000000","tz":120},"subject":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027","message":"docs: hw_rng_model: Document that it has no effect since \u0027Ussuri\u0027\n\nAlthough, ``hw_rng_model`` is no longer useful, as noted in the Nova\nChange-Id: I28e66c9640c38d23b8c0dbd0b05f5260bfcf6d30 (\"libvirt: Add a\ndefault VirtIO-RNG device to guests\"), we may extend it to take further\nvalues to enable other use cases (details in the Nova commit).\n\nHence we should not yet deprecate it.\n\nChange-Id: Id24ff7da83ffb8a27767f5f74b12878ecf7bb426\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89f3fd694064bb0ba17f14d868d04e733372929b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89f3fd694064bb0ba17f14d868d04e733372929b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
