)]}'
{"id":"openstack%2Fglance~704882","triplet_id":"openstack%2Fglance~master~If82d9b396afd87c65e035868b595ff5683cb19f6","project":"openstack/glance","branch":"master","topic":"bug/1861334","hashtags":[],"change_id":"If82d9b396afd87c65e035868b595ff5683cb19f6","subject":"Use default cors config also when run as wsgi app","status":"MERGED","created":"2020-01-29 20:29:56.000000000","updated":"2020-02-19 14:01:11.000000000","submitted":"2020-02-19 13:58:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704882-1582120723401-25e6796c","meta_rev_id":"a06c909e90efcaaf09102c8995b6824f5964c9ac","_number":704882,"virtual_id_number":704882,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-19 09:06:18.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 13:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-14 14:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-02-19 04:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-02-19 04:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-02-14 10:54:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-30 07:51:44.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-01-30 07:52:43.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-02-06 08:58:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 07:51:44.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-30 07:52:43.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-02-14 14:23:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-19 04:46:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-02-19 09:06:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-19 13:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d6d42e78e5fa7e7b8a2acdb329f693e1a41e4db","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-29 20:29:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e724ecbde97e106d9209e3d8af8202d49cb6d1cf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-29 20:30:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3c107f4b9510c9c95fb62bc166a619b98460b059","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-29 20:31:09.000000000","message":"Patch Set 1:\n\ntested in https://review.opendev.org/702132 (will be red but glance should pass in chromium)","accounts_in_message":[],"_revision_number":1},{"id":"91b0af85ddde785cdcabe181cdf7937e171cbcec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-29 20:31:50.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/82/704882/1/check-not-vote/ext-glance-zuul/2d1f341 : FAILURE in 1m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7818e48b6f02cd41e0fedd898c4057f4bfc71ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 21:58:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aafc8c3a21aa4c27ae862cfdb145c490 : SUCCESS in 1h 11m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b25f92fa0a54d1eacf9b82cc56556f2 : SUCCESS in 1h 22m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eddef3409c945cabc92aca54540eb8d : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f654ca73f94485f8679ba116cc9ee54 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0efe9f4ea324d7fa5494a868504faf0 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bbdbcf8da884f0187ecd0eff965b975 : FAILURE in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b81c9fb3679b4860ab947ebde689a95f : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007530cad3bc48309fef3395b65f6ec9 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a2d9712aed40b685796d408d15161f : SUCCESS in 10m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0668272fff84426876068a7df01261b : SUCCESS in 11m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/69d0770ca3784fb583bf9b007678fe3b : SUCCESS in 11m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ff269ce81aa84e56ae1dc8750df59d59 : SUCCESS in 4m 50s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9932681bce14b8495c379191198b733 : SUCCESS in 1h 19m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0f463807794496f9e306af3feff52df : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":1},{"id":"57dee93268f6dc68b6b5eb101f2f48c3ae66732a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 07:51:17.000000000","message":"Patch Set 1: -Workflow\n\nrecheck\n\nit works","accounts_in_message":[],"_revision_number":1},{"id":"efdddf9aea741cc55e49c3346739fbe5c0d6ea47","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-30 07:53:12.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/82/704882/1/check-not-vote/ext-glance-zuul/e8df475 : FAILURE in 1m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1c533e29936d5728d26bb225e477390592fb430a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 09:21:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c8e2e7e4dca94207aa4d734d918544e6 : SUCCESS in 1h 04m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aaa6cc2a866d47e7a76b3d277af7f305 : SUCCESS in 1h 25m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb7b57a241fa4877874400df168a2ba8 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1821e37d55493fa6e9b9835b090c64 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd43d4a69fe047da83a207d309ab0dba : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9731cd2cdaca4de09968b50d43cb2351 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/653f0e32089f4656a05a110f29b1e236 : FAILURE in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc1a0032e9140078d4b033183007f08 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4cb46e3878f416ab5d1bc328dc67766 : SUCCESS in 12m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/376dee3a94714ab68d42bd6b15198c5e : SUCCESS in 15m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0be537fc201343938d67c0a27865e9d0 : SUCCESS in 15m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/727f6c5ce7f14b3a8bf957896cff26c0 : SUCCESS in 6m 37s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83ac22882d4d405baadcbf17a99df99a : SUCCESS in 1h 24m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ea762a07fa54eec81702583e60b0f3b : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0d2625a38212b57f6624c0c180230ab2abd726dc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-06 08:58:07.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"4a2150a04698eb17d67bba600350148765950cba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-14 10:54:24.000000000","message":"Patch Set 1:\n\nbumping, it\u0027s just one line, folks :-)","accounts_in_message":[],"_revision_number":1},{"id":"df849dbbfad0f63a840904d19908c7a6f5cc0af4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 14:23:59.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks simple enough.","accounts_in_message":[],"_revision_number":1},{"id":"73fc15ce9e0161ab7f2c652734eebd7c3ce180b9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-19 04:46:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41246b91d4a49a6d944523f401a1866ea0628778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 04:46:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b15cf1957104cc61a076c971ea525cbce618d21c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 06:46:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a3b075b10ee2427688b8901a7021f14a : SUCCESS in 59m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6af312b3392d4bc582f21b394c3f8840 : SUCCESS in 1h 38m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3f44ecea7414b73ae0ad12e9cfe2b2d : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57accd709fc42a9aa92f90d43bf0d39 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9d0c2b5f4b14f9da0e1faa541b08c2b : FAILURE in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/514ab61b6058466ab9bd8fa4f173be38 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb37deb48584e7cbbe0ab59b5e311b8 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af2bfe1ba40b4bd6ac77817a9710eea5 : SUCCESS in 11m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e6be9e573d64b9180e5cdc849ecd228 : SUCCESS in 12m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1fb244b9d8d74d7baa62ec08987e2c5a : SUCCESS in 11m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f94a561e8f24e4f9bf9ed8d3e48aa23 : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"57dc5ff2fe9dcd23107d4dc8e2fba12dc77cd187","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-19 06:50:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bb35e1776df4049ffe70f7a107f91435191b0eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-19 06:52:52.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/82/704882/1/check-not-vote/ext-glance-zuul/4c32eff : FAILURE in 1m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a3e325c873ed43b48e92338489e19381a83e0e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 08:56:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d237f6fcb7ff472887bb751a3922fd2c : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b5830c5ef1a4e12bf72b6b5d547ec38 : SUCCESS in 1h 32m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3e92719993243958f97e644ed57e991 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cef92cb19bd4342b910fc53dd32464d : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26a3a1f11aae4e45a748306cceedb614 : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddc7e8b53bdf4bc1af063c8fd49ea2df : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58c4471c3c5c44f1871d26e0fddbbf91 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e2c003b1c24950879c5b0e1b8ddcfd : SUCCESS in 10m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6545506b572841f2b50188571eee9a69 : SUCCESS in 14m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/278d482471ec4e49937de4d57cef58fc : TIMED_OUT in 31m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f3407d3817e34baebc877494652f374a : SUCCESS in 12m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5fb1efb4b25a4e939d4feb762f0d8180 : SUCCESS in 6m 22s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fe633dec1b24a3a84d059ce07eebe8b : FAILURE in 14m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0ab63e850ee40f89877d823d74ec6ad : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"36eb965cd80da07a5583d6f7d63fa3b38ec8652d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-19 09:04:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ade4048c3718d6230c41c5fe3ed361cd1f442d06","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-19 09:06:18.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/82/704882/1/check-not-vote/ext-glance-zuul/b0fd2bb : FAILURE in 1m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e1809bb7e63cfa9dfcabfb41f4abea849126d357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 11:20:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d08dbc855862445cac61b0b899dde2e3 : SUCCESS in 1h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea280de69ff94e619f8c304141e99885 : SUCCESS in 1h 23m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8662df542c2f4b27821d3b8c18c3fcf9 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b608a2cff1b543e88e78eafc402960ae : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d358482ebb4129a07ee323d404ac7b : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5592d2b96f1d4b06912f4ddd1f6a1180 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85a67c28b58e409bacc2b2b7f9c24395 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328853f12f0740a582ec4929de61e622 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45eacaeefed94f5a98d0b7eec1329402 : SUCCESS in 10m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b33b03379150435e973755f5dad10eee : SUCCESS in 8m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1f9fec255bb5453aa523440d17312b5d : SUCCESS in 11m 44s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/21b6e76b89c14e599dab0db9b0b72540 : SUCCESS in 5m 56s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a6572db6f2c04a83a49344acc65522a2 : FAILURE in 14m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f2bdf07cd4547f99834fa0364aa3b7d : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5e28afff536217f193c5619deb7a63209c18016a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 11:20:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2376ced75cf3f356b0c7967f679ba80fc85a87e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:58:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4f9b4e92024914d5469fad2bba95d57803c91aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:58:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77c5696e267b4fd89daa44e02e0f6a7e : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1cddd3cdeb954d04b90ed8f348d15953 : SUCCESS in 1h 16m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaaf1eb447024d8898b1408a4ce7674c : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6e0f0a243d4b51975a688717e31b5f : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84420ad58b1142d5a17a77051a60ae5f : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d17c74babb244fd80779d8e3f69c9d7 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b026921769de438497c396b4e22f2501 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ebea13855c14666b9930aa6cd732417 : SUCCESS in 11m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51a8d2bc7bbb43f7a34d439c15e09022 : SUCCESS in 11m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/528da6aa7cba46e2b11e2fbae8c8bd1f : SUCCESS in 9m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2c58b0a3d54443aad1a38121418755e : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"63042ea5fa884de428b2ef812f3297bb9ccdea1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 14:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1083f41fcaf54e85b8fea8f726ba4474 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e14d3ac4e354509bc05fe24cd32b392 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a6fcc2ea52f38ab971551734244e45a91e5b074","revisions":{"9a6fcc2ea52f38ab971551734244e45a91e5b074":{"kind":"REWORK","_number":1,"created":"2020-01-29 20:29:56.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/82/704882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/704882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/704882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/704882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/704882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/704882/1"}}},"commit":{"parents":[{"commit":"26c5c8dff670617ecbcc2b319b25e632fa88cb4a","subject":"Merge \"Remove OS::Compute::Trust metadef\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26c5c8dff670617ecbcc2b319b25e632fa88cb4a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-29 20:27:28.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-29 20:29:54.000000000","tz":60},"subject":"Use default cors config also when run as wsgi app","message":"Use default cors config also when run as wsgi app\n\nThis fixes Glance CORS behind uWSGI as used in devstack.\n\nChange-Id: If82d9b396afd87c65e035868b595ff5683cb19f6\nCloses-bug: #1861334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a6fcc2ea52f38ab971551734244e45a91e5b074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a6fcc2ea52f38ab971551734244e45a91e5b074"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
