)]}'
{"id":"openstack%2Fglance~705229","triplet_id":"openstack%2Fglance~master~I5a1c61430879a910e7b6c79effba538431959d56","project":"openstack/glance","branch":"master","topic":"bug/1861501","hashtags":[],"change_id":"I5a1c61430879a910e7b6c79effba538431959d56","subject":"Ensure store ID parsed from URI is properly encoded","status":"MERGED","created":"2020-01-31 14:46:01.000000000","updated":"2020-01-31 18:18:30.000000000","submitted":"2020-01-31 18:10:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705229-1580494219631-fa858c1e","meta_rev_id":"e4048cc6ad78a2dfbfc8b7b5b4d739993c2e79aa","_number":705229,"virtual_id_number":705229,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 18:10:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-01-31 15:41:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-01-31 16:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-01-31 15:41:19.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 15:41:19.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-31 16:54:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-01-31 18:10:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d0467103a22cad60e10f9c11b592d3d7a1464cc","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-31 14:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68afac7a5314542445cb8920c5a8abff891d7ab5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-31 15:41:04.000000000","message":"Patch Set 1: Code-Review+2\n\nSimple and clear change, will ninja approve once zuul finishes the job.","accounts_in_message":[],"_revision_number":1},{"id":"f79badb69b802b778b16588102efd07f76e68d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 16:13:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/90ad2612a8324df8bfd49acaee13b45f : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc4cc5299fa44363a3cfa6de77ff338b : SUCCESS in 1h 21m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5b03059583c4ef5b6a719c6a7415e5b : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6336c43fb8e4665a392d4e628a0d7a1 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cd09afa1c174b7596ea37da7f069676 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e27686b5476f4cc7b9637ada52360407 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5729029a16ed4786be542123a3489fa7 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c2d08542d5d43f783c57230746351c7 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/52e7ba6175734971ae1d70f89acfca7a : SUCCESS in 9m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/932b76a06a1e484196b49e3e03235083 : SUCCESS in 11m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3cb3bc852ee3440c87729d12300d8437 : SUCCESS in 5m 26s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/41e410a3f0684c16b60fa94be90e10bf : SUCCESS in 1h 10m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05e4cf0367804657aefc02eb21686c99 : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"19126e93fd7bb77aa1d470ec62706004182b65de","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-01-31 16:54:30.000000000","message":"Patch Set 1: Workflow+1\n\nSimple change, Ninja approving it, so that we can backport it to Train.","accounts_in_message":[],"_revision_number":1},{"id":"ac1f7549d1ebf5937d32531162ce248917535202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 16:54:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa942bf6daedcb466e4f72219536ae0736eceec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:10:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da615c6593dcba6bfaede2dde82edde862e5576e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:10:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a7467fd0430411585b8f0ff4825df66 : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee357051be24465493b32be9bfab8014 : SUCCESS in 1h 13m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90841a83e969446b82394c2ca041901e : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e9745aaff1405a88b5172585bb33c4 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1db2bc1c6b444d9c30da72ee4297ba : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48f6cd164a464d29b366bd69574f632e : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7347adfc331644a1ab3a588b33aa2768 : SUCCESS in 9m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6702da5dfb95438f9d4a6282a0bdde48 : SUCCESS in 10m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d569e8077f1b4da394a503e029b71051 : SUCCESS in 10m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87c87fdbf65c416e9d9321c9c1057e8d : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0ab616e3250aee084f4946160d7d5184e886b3f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:12:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fda786345ea746179926b1a256b0cb16 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"673a60b57e0c63bb213ebc3512571051ff136469","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2020-01-31 18:18:30.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c5bdc5536690f5b3ae44c4d9ae2e96470287eb56","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8649fdc2a293009899d2aa918ec360e30adc8e77","revisions":{"8649fdc2a293009899d2aa918ec360e30adc8e77":{"kind":"REWORK","_number":1,"created":"2020-01-31 14:46:01.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/29/705229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/705229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/705229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/705229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/705229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/705229/1"}}},"commit":{"parents":[{"commit":"26c5c8dff670617ecbcc2b319b25e632fa88cb4a","subject":"Merge \"Remove OS::Compute::Trust metadef\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26c5c8dff670617ecbcc2b319b25e632fa88cb4a"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-31 14:25:31.000000000","tz":-480},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2020-01-31 14:45:19.000000000","tz":-480},"subject":"Ensure store ID parsed from URI is properly encoded","message":"Ensure store ID parsed from URI is properly encoded\n\nEnsure the store ID parsed from a request URI is unicode. This fixes\na py27 issue that allowed an incorrectly encoded value to be added to\nthe image\u0027s metadata.\n\nSupport for py27 is dropped in Ussuri, but this fix is required on\nearlier releases that do support py27.\n\nCloses-Bug: #1861501\nChange-Id: I5a1c61430879a910e7b6c79effba538431959d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
