)]}'
{"id":"openstack%2Fglance~705526","triplet_id":"openstack%2Fglance~master~I4298bb835ec60b79a0010e52318ada19eb24ad80","project":"openstack/glance","branch":"master","topic":"bug/1861723","hashtags":[],"change_id":"I4298bb835ec60b79a0010e52318ada19eb24ad80","subject":"Configure Glance TCP socket after the store initialization","status":"MERGED","created":"2020-02-03 19:19:54.000000000","updated":"2020-02-13 14:33:37.000000000","submitted":"2020-02-13 14:25:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705526-1581603913908-5f1f21a0","meta_rev_id":"4e8df1e3529b597c9d9a1b96828b146cdbf69610","_number":705526,"virtual_id_number":705526,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 14:25:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2020-02-13 12:33:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-02-04 23:46:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2020-02-04 04:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2020-02-13 12:33:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 04:34:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-02-04 23:46:13.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2020-02-13 12:33:43.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-02-13 14:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32e1bec7082e813e3fc1bbf17db4af2235577d07","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-03 19:19:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cfb30dbf48358928496e217596b4d99434751cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 23:01:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/53ea4820287c478baa5a067d8c901325 : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba4ba2a181d94c3b8cca2dc2d524d25b : SUCCESS in 1h 24m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f67af3636a64303b905537a375b97fd : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33300286810a402da081e3c53a315e11 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d3cc24fcf524cbc91213686f268b285 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e4ca2069f9545e085959c78e12a52a6 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b09c9816c2c2469eb4b8034944e8cb9e : FAILURE in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ce74490beb4d7cb9c313116125cbef : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/678e5e95f3a9417993f655c7862f6a2b : SUCCESS in 9m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d3b4b027286e469cb596794e533ca48a : SUCCESS in 9m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/25d8f2bf8d534f9fb8fbc372df4bbe79 : SUCCESS in 4m 11s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4278d5a95a7842b3ad304351a6481ad4 : SUCCESS in 1h 10m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f0694de3e88482186f45fe629074674 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"152cfd25ed10f210f178e223e566d918644e2ccd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-04 04:34:36.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense, thank you for working on this!!!","accounts_in_message":[],"_revision_number":1},{"id":"938953d0623b02d4a997587f9abaffd7ac2c8e7c","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2020-02-04 23:46:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ca4d0775a2fb822243d4b1b51c46e1711132d54d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-13 12:33:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNice catch!\n\nThanks for your help fixing this.","accounts_in_message":[],"_revision_number":1},{"id":"7b4eb8c19622ea17661122192acfc2a6522f69e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:34:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b939264903ec2638314b543e293d20621a57b9f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:25:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e1a96ace58f43f67c5d7e4d08923b4ec9a12ee7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:25:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ea9e075335834252a177ed6b932a1d71 : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eed704f7e69a4adda80a3690dd560a12 : SUCCESS in 1h 16m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/197c5da24dcc420ea8024ea555be0953 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdef1b2a1dd468b920ca5d945e1ab3b : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcd9dd40d6d24ebf9a219d7304afff54 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da17a247c33941d0ba93d5a6ee8029bb : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ba110c7a464e60bd258c6ede292a81 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90721082c5c8460b8e57d0693c85efcb : SUCCESS in 8m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/16249dbc02ed49c490c5d69880c95cf9 : SUCCESS in 8m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/435cebaa48994101ba6e59687d8ea50c : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"471479bacf61f6428b19c23b6471113cceee4532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:27:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42cf7b33015e4221aa6f74506513e5fe : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e570fd4fd79c77e8b9fd972440fec6f50563b6c4","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-13 14:33:37.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a1fd2b4de576b0bcb718a042cf257ce819543dbf","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9902c26a1f75149793b92d92968c40288f3d9ae","revisions":{"d9902c26a1f75149793b92d92968c40288f3d9ae":{"kind":"REWORK","_number":1,"created":"2020-02-03 19:19:54.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/26/705526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/705526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/705526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/705526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/705526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/705526/1"}}},"commit":{"parents":[{"commit":"8649fdc2a293009899d2aa918ec360e30adc8e77","subject":"Ensure store ID parsed from URI is properly encoded","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8649fdc2a293009899d2aa918ec360e30adc8e77"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2020-02-03 19:19:38.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2020-02-03 19:19:38.000000000","tz":180},"subject":"Configure Glance TCP socket after the store initialization","message":"Configure Glance TCP socket after the store initialization\n\nWhen multiple stores are being used with rbd backend, Glance tries\nto get fsid of the cluster using rados library.\nIf even one of the rbd backends is unavailable, glance-api wsgi fails\nto start properly, but continue listening on the TCP socket and is not\nresponding at any request until the timeout.\n\nThis change moves socket configuration after the store initialization\nto make sure that failed Glance API process will not be\nconsidered as up by the haproxy TCP check.\n\nChange-Id: I4298bb835ec60b79a0010e52318ada19eb24ad80\nRelated-Bug: #1861723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9902c26a1f75149793b92d92968c40288f3d9ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9902c26a1f75149793b92d92968c40288f3d9ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
