)]}'
{"id":"openstack%2Fglance~706942","triplet_id":"openstack%2Fglance~master~Icbe52793fd33ba61944ee59bbfa538003ffbfee9","project":"openstack/glance","branch":"master","topic":"specs/copy-existing-image","hashtags":[],"change_id":"Icbe52793fd33ba61944ee59bbfa538003ffbfee9","subject":"WIP: Copy existing image in multiple stores","status":"ABANDONED","created":"2020-02-10 17:53:07.000000000","updated":"2020-02-12 06:15:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e924ed7c80d923c171dc9dc1d323bc96a0180b5","_number":706942,"virtual_id_number":706942,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-11 00:52:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2020-02-10 19:30:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 00:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0398b7a1d45370ac996a364568d797293bf36b47","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 17:53:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f308633c5f48b35b4231896c1802898b45e67a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:53:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027gate\u0027][\u0027jobs\u0027][2][\u0027tempest-integrated-storage\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py37\n  \n      gate:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py37\n          - tempest-integrated-storage:\n          - grenade-py3:\n          - tempest-ipv6-only:\n      experimental:\n        jobs:\n          - glance-tox-py37-glance_store-tips\n          - glance-tox-py36-glance_store-tips\n          - glance-tox-functional-py37-glance_store-tips\n          - glance-tox-functional-py36-glance_store-tips\n          - barbican-simple-crypto-devstack-tempest\n          - glance-dsvm-grenade-multinode\n          - devstack-plugin-ceph-tempest-py3\n          - tempest-pg-full\n          - tempest-full-py3-opensuse15\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-py37-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py37-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-py37-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py37-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py37-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py37-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-py37-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 222, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6871a9e1a46224cc5d54cc00c07059a6a0dff7a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 17:54:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6557e455c763470a3acd1a5dd0fecdb7ef3988c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-10 19:30:05.000000000","message":"Patch Set 2: Workflow-1\n\nTesting patch to confirm py37-functional timeout problem","accounts_in_message":[],"_revision_number":2},{"id":"2f410451fc1142114710d15caba8b1aeba6476b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 00:52:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/319f2aa58ea540a0a8be16db57f559bf : SUCCESS in 4m 17s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8d6d4c342dc243babc495fb81a32b0a3 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d08c61e9bc674ca0a6a33ece4c9efbdd : SUCCESS in 10m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/73ee4216fa3c4c2e9526d7ced4665375 : TIMED_OUT in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5e8669206ff4fa7664902ce91643bf3f6d54c2b4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-12 06:15:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b08f6af2ed3eeb99c29604a8b08aebb4501f1185","revisions":{"74a08bcb57125f284dc06d44c9bd06b08c76e2ff":{"kind":"REWORK","_number":1,"created":"2020-02-10 17:53:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/706942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/706942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/706942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/706942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/706942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/706942/1"}}},"commit":{"parents":[{"commit":"92492cf50461e214b777c707148886a8e87f340d","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 17:51:54.000000000","tz":0},"subject":"WIP: Copy existing image in multiple stores","message":"WIP: Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nChange-Id: Icbe52793fd33ba61944ee59bbfa538003ffbfee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74a08bcb57125f284dc06d44c9bd06b08c76e2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74a08bcb57125f284dc06d44c9bd06b08c76e2ff"}]},"branch":"refs/heads/master"},"b08f6af2ed3eeb99c29604a8b08aebb4501f1185":{"kind":"REWORK","_number":2,"created":"2020-02-10 17:54:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/706942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/706942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/706942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/706942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/706942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/706942/2"}}},"commit":{"parents":[{"commit":"92492cf50461e214b777c707148886a8e87f340d","subject":"Add ability to import image into multi-stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/92492cf50461e214b777c707148886a8e87f340d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 11:05:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-10 17:54:50.000000000","tz":0},"subject":"WIP: Copy existing image in multiple stores","message":"WIP: Copy existing image in multiple stores\n\nAdded new import method \u0027copy-image\u0027 which will copy existing image into\nspecified list of stores. Introduced additional task which will serve\nas internal plugin which will allow copying existing image into staging\narea and then this data will be uploaded to specified stores via regula\nimport flow.\n\nNOTE: This new import method \u0027copy-image\u0027 is only supported if multiple\nstores are enabled in deployment.\n\nChange-Id: Icbe52793fd33ba61944ee59bbfa538003ffbfee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b08f6af2ed3eeb99c29604a8b08aebb4501f1185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b08f6af2ed3eeb99c29604a8b08aebb4501f1185"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
