)]}'
{"id":"openstack%2Fglance~707759","triplet_id":"openstack%2Fglance~master~I1e90ceee1f87291668e447f180f37bb809763836","project":"openstack/glance","branch":"master","topic":"devstack-breakage","hashtags":[],"change_id":"I1e90ceee1f87291668e447f180f37bb809763836","subject":"Revert \"Remove all example configs for deprecated registry\"","status":"MERGED","created":"2020-02-14 03:30:56.000000000","updated":"2020-02-14 04:41:26.000000000","submitted":"2020-02-14 04:40:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707759-1581655202295-3390009c","meta_rev_id":"eafb50986b3acab300c636f95b72267ffa61a749","_number":707759,"virtual_id_number":707759,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 04:40:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-14 03:35:08.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-14 03:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-14 03:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 03:45:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-14 04:40:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92b451bbfeb2f78ab4706920ec74b1c4d6aa2d4a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-14 03:30:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"153c88aa2dad98a64dd9ec26d5937cba233834f8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-14 03:34:51.000000000","message":"Patch Set 1:\n\nThe immediate issue is \n\n 2020-02-14 00:14:50.764213 | controller | + lib/glance:configure_glance:201          :   cp -p /opt/stack/glance/etc/glance-registry-paste.ini /etc/glance/glance-registry-paste.ini\n 2020-02-14 00:14:50.765764 | controller | cp: cannot stat \u0027/opt/stack/glance/etc/glance-registry-paste.ini\u0027: No such file or directory\n\nI had a quick look and this seems to remove a number of files used in various ways in devstack, so it\u0027s not as simple as pulling out that one copy.  If you could put this back with a depends-on a devstack change that would be great.","accounts_in_message":[],"_revision_number":1},{"id":"30e8e08218497c551c71df554404f41339db2977","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-02-14 03:35:08.000000000","message":"Patch Set 1:\n\ne.g. https://zuul.opendev.org/t/openstack/build/de205d64d85a4106be6bb299110587c7/log/job-output.txt","accounts_in_message":[],"_revision_number":1},{"id":"bfd61449202a8893df7839edb77b16584c836ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 03:43:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f6e5030e6d24af5b966164905bed53c : SUCCESS in 3m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4c0d590bdd74f16ae5a663e9cfe9ed8 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445d91087b504b67b1b89a433e23210d : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce3d4eeae4134198ac6dc08ec7b260ed : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4aa074ffaeea44bd8200c57455a54f85 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33ce2fef818d4afb8406d15d866f33a3 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e007ccb81234c6fae7e6f7d2bd8203e : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de052983ce25403abca69554250e43ed : SUCCESS in 8m 22s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b361d1fa4c754b05a18446bdc8fbf792 : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fc5735ba8f9ada4ed72cc312bd83fddc257fb12b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 03:45:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPushing through to unbreak things.","accounts_in_message":[],"_revision_number":1},{"id":"ffba743cfc026a09bd78763804a967f2765fb132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 03:45:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c6ec72d3552204252b37bafd1c9d0b4e45546eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:40:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2dd696c41e39eef30b23d267686ed8111c28df90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:40:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db504d1168474f4cbfcd2482f22712c0 : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65bc24502c6f4ea99e8fbb6f6cad3c35 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c5ba49df06f4a8c88d926495cc48971 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61fa469614ee4471936a96d3342b30ee : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb47fe26a4034b5498ae7049336e5828 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8aff535c9c49bdba44e6c98e8bff34 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d83cf291c960424da4cc8f339fe8eea7 : SUCCESS in 9m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/51ec4ddf1bfa4aab9256753b8b06aad3 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3b8f7180074fe7a9257085fb86bd330dc5a1ecae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:41:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e099ed73b74780b17e111d7e824b14 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0035ab253819246491f11ffeecdab99e78d75629","revisions":{"0035ab253819246491f11ffeecdab99e78d75629":{"kind":"REWORK","_number":1,"created":"2020-02-14 03:30:56.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/59/707759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/707759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/707759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/707759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/707759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/707759/1"}}},"commit":{"parents":[{"commit":"2a28696de9e18a3866631507739944ceb3460872","subject":"Remove all example configs for deprecated registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a28696de9e18a3866631507739944ceb3460872"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 03:29:46.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-02-14 03:30:44.000000000","tz":660},"subject":"Revert \"Remove all example configs for deprecated registry\"","message":"Revert \"Remove all example configs for deprecated registry\"\n\nThis reverts commit 2a28696de9e18a3866631507739944ceb3460872.\n\nDevstack still references a number of these files; revert until\ndevstack incorporates removal.\n\nChange-Id: I1e90ceee1f87291668e447f180f37bb809763836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0035ab253819246491f11ffeecdab99e78d75629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0035ab253819246491f11ffeecdab99e78d75629"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
