)]}'
{"id":"openstack%2Fglance~707811","triplet_id":"openstack%2Fglance~master~Icf622bce0d1393c37a7f7a176050750374d2cefe","project":"openstack/glance","branch":"master","topic":"test-zuul","hashtags":[],"change_id":"Icf622bce0d1393c37a7f7a176050750374d2cefe","subject":"Revert \"Revert \"Remove all example configs for deprecated registry\"\"","status":"ABANDONED","created":"2020-02-14 10:16:53.000000000","updated":"2020-03-17 08:40:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed597ac037338d7368b286443bdea4ef82fc5a30","_number":707811,"virtual_id_number":707811,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-14 11:39:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 11:39:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f766f89203f9382ec58dbc538e93955e11611780","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-02-14 10:16:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75ac0284737b789f1a1a43d664122a7e5988d57b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 11:39:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a68b194d9b2e4cfb8838c4d386a69283 : SUCCESS in 4m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2324d92bcb884400aa4a6b635b94dfe4 : SUCCESS in 1h 01m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcf84c61445547b797b48a66e30e1027 : FAILURE in 17m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/960f057e4da44b6b8a79f05139e350bb : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfbde859e0bc4bbf8995351caed19b40 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb71d42854014b2f8e779c977f373eae : SUCCESS in 8m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58e0d3f3a3744a33bfc1ad9a40d5667b : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd5f54890034dc5a519d020fff9cc5c : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd60c801001e429fae436fe03fcc589e : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8473110a61b7431480f4dd79d2bcc0fd : SUCCESS in 8m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/56879817be0b423b9ca652f05ca8794c : SUCCESS in 12m 28s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/794759a9f70f40b496ab66c95bd287a3 : FAILURE in 13m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e133bf057e924ff5a7a3b4b435573799 : FAILURE in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f66518bfd49edaeba15437f6aa74c6dcd61b20df","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-17 08:40:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ac31fd98ff20cc5f6aa4a50da2b5808efda7e96","revisions":{"2ac31fd98ff20cc5f6aa4a50da2b5808efda7e96":{"kind":"REWORK","_number":1,"created":"2020-02-14 10:16:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/11/707811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/707811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/707811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/707811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/707811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/707811/1"}}},"commit":{"parents":[{"commit":"9181935d6c86ad6a0005720bc9638164b8e619be","subject":"Amend tempest-irrelevant-files to avoid false positives","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9181935d6c86ad6a0005720bc9638164b8e619be"}]}],"author":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-02-14 10:08:37.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-02-14 10:16:42.000000000","tz":0},"subject":"Revert \"Revert \"Remove all example configs for deprecated registry\"\"","message":"Revert \"Revert \"Remove all example configs for deprecated registry\"\"\n\nThis should now fail on devstack-based jobs, allowing them to be fixed\nbefore this patch gets merged again.\n\nThis reverts commit 0035ab253819246491f11ffeecdab99e78d75629.\n\nChange-Id: Icf622bce0d1393c37a7f7a176050750374d2cefe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ac31fd98ff20cc5f6aa4a50da2b5808efda7e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ac31fd98ff20cc5f6aa4a50da2b5808efda7e96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
