)]}'
{"id":"openstack%2Fglance~709109","triplet_id":"openstack%2Fglance~master~I10e1e738aeacc708350993f05d373eaa1293fd27","project":"openstack/glance","branch":"master","topic":"bug/1850412","hashtags":[],"change_id":"I10e1e738aeacc708350993f05d373eaa1293fd27","subject":"Document os_admin_user in Useful Image Properties","status":"MERGED","created":"2020-02-21 11:25:56.000000000","updated":"2020-02-24 17:34:52.000000000","submitted":"2020-02-24 17:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709109-1582565231332-eaa21e4c","meta_rev_id":"8202ecb4814249c866c3e8962e7ff4826b793539","_number":709109,"virtual_id_number":709109,"owner":{"_account_id":15279,"name":"Russell Tweed","email":"rtweed@redhat.com","username":"tedlit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 17:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2020-02-24 10:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2020-02-24 10:28:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 10:28:55.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-02-24 17:27:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec613a387001dea66c622e3a6a833336aca7b464","author":{"_account_id":15279,"name":"Russell Tweed","email":"rtweed@redhat.com","username":"tedlit"},"date":"2020-02-21 11:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e5de4bce5665bf2e6949ab5c50cf236dc1b5764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 11:37:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c4b6112cf445528c950bc9d209e6a2 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68a0915e51f498eba743cd4978639e8 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d20c5eec893d166bcbc482b3295008359219fc53","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-24 10:28:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Russell,\nThis looks good to me.\n\nAs per Glance review policy I will ninja this doc-only patch in right away.","accounts_in_message":[],"_revision_number":1},{"id":"8ab5a317a9645f9b410f5c8b7a7c55ee4e89af7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 10:29:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa2609dab80cefe5bbb11eac31879680f0d0cb8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:27:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c898c747304ff9895dc711d9828d7e : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fbba5c66594512aec5572903cc0a6b : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2f8a02efc1c658216cfad8ee003cbe18174ac0bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:27:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8c3a402e124c23b3a6e25690c5e0da37656991a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:34:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b085c255d0f4aedaf9d888da9c812bb : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"517ff4fbe4694adc95460ca209e2229a3255edb8","revisions":{"517ff4fbe4694adc95460ca209e2229a3255edb8":{"kind":"REWORK","_number":1,"created":"2020-02-21 11:25:56.000000000","uploader":{"_account_id":15279,"name":"Russell Tweed","email":"rtweed@redhat.com","username":"tedlit"},"ref":"refs/changes/09/709109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/709109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/709109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/709109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/709109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/709109/1"}}},"commit":{"parents":[{"commit":"2456c7d0000e8e3900e5812a820bbc31bc4d1cf1","subject":"Merge \"Use default cors config also when run as wsgi app\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2456c7d0000e8e3900e5812a820bbc31bc4d1cf1"}]}],"author":{"name":"Russell Tweed","email":"rtweed@redhat.com","date":"2020-02-21 11:22:44.000000000","tz":0},"committer":{"name":"Russell Tweed","email":"rtweed@redhat.com","date":"2020-02-21 11:22:44.000000000","tz":0},"subject":"Document os_admin_user in Useful Image Properties","message":"Document os_admin_user in Useful Image Properties\n\nAdd documentation of the os_admin_user property to the Useful Image Properties guide.\n\nChange-Id: I10e1e738aeacc708350993f05d373eaa1293fd27\nCloses-Bug: 1850412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/517ff4fbe4694adc95460ca209e2229a3255edb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/517ff4fbe4694adc95460ca209e2229a3255edb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
