)]}'
{"id":"openstack%2Fglance~709647","triplet_id":"openstack%2Fglance~master~I2164bfe0ffb144223336cf6c9d59f89a5de8ef8f","project":"openstack/glance","branch":"master","topic":"devstack/glance-multiple-store-support","hashtags":[],"change_id":"I2164bfe0ffb144223336cf6c9d59f89a5de8ef8f","subject":"Define glance multiple store integration job","status":"ABANDONED","created":"2020-02-25 01:52:30.000000000","updated":"2020-08-16 15:57:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2612093357b9a6105ac3c813d4cc07a58ef6dc81","_number":709647,"virtual_id_number":709647,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"date":"2020-02-26 10:41:28.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"date":"2020-03-18 14:59:41.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-26 12:20:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2020-02-26 10:49:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 10:41:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-02-26 10:49:30.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-02-26 12:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f197398f4d8a79b60b509d4bddc70e830c5d215","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-25 01:52:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d96d8751678155d9a72942ae9bfa17a9b31dfe52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 03:25:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/545c55fb7e834ced9595a6aac417ae0e : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3972e2da84324e529b5fc27f1fb488d5 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0508c1627a6c4bafbed7110ef81d74ad : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/862e9c69837d48789d2e3c578aa93bab : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca3829fbd01e4224af2b2a3c35053117 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153197f380e74d3da2bcf5fa7d95e471 : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50a82a745f7b4db09484fe25902107f1 : SUCCESS in 8m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/10713c6b1d634552a02fcd3ca1c360c3 : SUCCESS in 9m 07s\n- glance-multi-store https://zuul.opendev.org/t/openstack/build/c638d28b7c164e4bbfa2f612f15caac6 : SUCCESS in 1h 27m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fcc0959fb67755843942c4d5e5b8de2b6ed5f7a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-25 05:59:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7aba57b2d88ba8a7d27483696548aa5055e3e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 07:38:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec5c47c3c56d4754a8f9cb2a7f6bef8e : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b43c3b09536461e821a80713e9f0d04 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7abacf2e5b664bc180919c6d6170ce23 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64e1d4cf7d2440f78cc35b5bb00a7f5d : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86594df06f3e4593ac4d282144666720 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d6fc51e15d4b4f9962d1f8d302f5bc : SUCCESS in 8m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45a943691c6744ecb842bc2cccd31490 : SUCCESS in 11m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/149d8c76709f4ca38e9fb8ae76d1730a : TIMED_OUT in 31m 34s\n- glance-multi-store https://zuul.opendev.org/t/openstack/build/e355c598e3cb4501a0715d2ea9cb8bc6 : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e2e4a2db0207d31aa21a93402e8e69fd65bfa609","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-25 07:40:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18f60a3639c83263240a806ac409dc1d5bf58c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 08:59:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/484be0f8a9f948f988b12d669793c40c : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb1d58828e143bb861de2973583b07a : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c04f9a26af64f289e2c81a49a3463be : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8f9840afdb2434dbb8c02116e908290 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c165129797f141afaa6d6191e8789768 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6063290e1d4315952b7a048a3d80ba : SUCCESS in 10m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/823fc3adeede42dfbcd1a9175874747c : SUCCESS in 8m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7d103a0955f5467bbfed04277adb7b06 : SUCCESS in 11m 29s\n- glance-multi-store https://zuul.opendev.org/t/openstack/build/d94812227c714ab0b837cb2003e939c7 : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9cdb8d2c3bc36d4ce786e2025870f6d62f6d7fea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-25 17:56:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c27f82492cc9db8ca3c36612c6b9f77715fdaa1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:02:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba9091f848284265afc7e4acfc80cb9a : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1bdf759237474eb093f894901f36e7 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b0877cdab24da89e95eb3bd118f762 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b3b7e533d9d45e48e88fbee7dac2888 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b6d9deef68640ccb5a7d608b703415f : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba97cf3794948dcaf1ef023fdd3b05c : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4deb9e288154bfd8cb38b89ee864b94 : SUCCESS in 12m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/72ad33a05a67458aaf77266235f37330 : TIMED_OUT in 31m 00s\n- glance-multi-store https://zuul.opendev.org/t/openstack/build/0854d3504a1a4334a6ba37cb3f844cd7 : FAILURE in 27m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7495903ce83d3449891bfeb18521022109031aed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-26 10:41:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ebf662f08a07cc049ad48e34f3a45318e44aae02","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-02-26 10:49:30.000000000","message":"Patch Set 2: Code-Review-1\n\nTo be able to do meaningful multi-store integration testing we need to fix the devstack deployment of glance first as majority of the multi-store features depend of taskflow and image-import.","accounts_in_message":[],"_revision_number":2},{"id":"b7c805617a48e778c181ce31bdf7ed0b4c1bae67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:20:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/411f15ed161b45e08244142b90215edf : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed0382fa72e0424bb7292625f26ba091 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25dc61ea0f104f89b4321c71fba765c6 : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ed434b348ae49f8853a38ea76e39eeb : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/091e82c8f2414b7083519d1ba9101f4e : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1fa5860334469f90d820a23163edb2 : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45f9f08bdc4245c88a3e590bd67a98b4 : SUCCESS in 8m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3092255297a4419bb20a2c30d4a92e09 : SUCCESS in 11m 31s\n- glance-multi-store https://zuul.opendev.org/t/openstack/build/2932694002704a4492b5a7aef84812d7 : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e1d7d947f3a3b2bbecba5003479cf9c010547eee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-18 14:59:41.000000000","message":"Patch Set 2:\n\n\u003e To be able to do meaningful multi-store integration testing we need\n \u003e to fix the devstack deployment of glance first as majority of the\n \u003e multi-store features depend of taskflow and image-import.\n\nYeah, this is to setup the infra for that. Depends-On devstack patch is also setting up the base setting in devstack.","accounts_in_message":[],"_revision_number":2},{"id":"dcc00e10b8b36b105d0c29b89327363c9ce32d0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-16 15:57:54.000000000","message":"Abandoned\n\nwe have job setup now","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a910853dadb0782d9fbb1747e6da94706aebdfe5","revisions":{"e771b430f4545dcdf47c2c9b83d0b2cb8150fc9a":{"kind":"REWORK","_number":1,"created":"2020-02-25 01:52:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/709647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/709647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/709647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/709647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/709647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/709647/1"}}},"commit":{"parents":[{"commit":"f46fe24616c8f338927b6490dcb23f110bd5d966","subject":"Merge \"Use devstack ceph plugin python3 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f46fe24616c8f338927b6490dcb23f110bd5d966"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-25 01:51:32.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-25 01:51:32.000000000","tz":-360},"subject":"Define glance multiple store integration job","message":"Define glance multiple store integration job\n\nDepends-On: https://review.opendev.org/#/c/689104\nChange-Id: I2164bfe0ffb144223336cf6c9d59f89a5de8ef8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e771b430f4545dcdf47c2c9b83d0b2cb8150fc9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e771b430f4545dcdf47c2c9b83d0b2cb8150fc9a"}]},"branch":"refs/heads/master"},"a910853dadb0782d9fbb1747e6da94706aebdfe5":{"kind":"REWORK","_number":2,"created":"2020-02-25 17:56:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/709647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/709647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/709647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/709647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/709647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/709647/2"}}},"commit":{"parents":[{"commit":"f46fe24616c8f338927b6490dcb23f110bd5d966","subject":"Merge \"Use devstack ceph plugin python3 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f46fe24616c8f338927b6490dcb23f110bd5d966"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-25 01:51:32.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-25 17:56:48.000000000","tz":0},"subject":"Define glance multiple store integration job","message":"Define glance multiple store integration job\n\nDepends-On: https://review.opendev.org/#/c/689104\nChange-Id: I2164bfe0ffb144223336cf6c9d59f89a5de8ef8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a910853dadb0782d9fbb1747e6da94706aebdfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a910853dadb0782d9fbb1747e6da94706aebdfe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
