)]}'
{"id":"openstack%2Fglance~711581","triplet_id":"openstack%2Fglance~master~Ib02387c4be70581df64cf5009750b4f30ae30159","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"Ib02387c4be70581df64cf5009750b4f30ae30159","subject":"Remove policy.json from setup.cfg","status":"MERGED","created":"2020-03-06 05:24:00.000000000","updated":"2020-03-06 09:34:33.000000000","submitted":"2020-03-06 09:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711581-1583487177423-74ea99eb","meta_rev_id":"49f73b375d81e35706d1d2ca9116e92410bc8c3c","_number":711581,"virtual_id_number":711581,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 09:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-03-06 08:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-03-06 08:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-03-06 05:24:43.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-03-06 05:25:37.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"date":"2020-03-06 05:25:03.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 05:24:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-06 05:25:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-03-06 05:25:37.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-03-06 08:12:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-03-06 09:32:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8007b046a42605f92ceed42d12210780ce637590","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 05:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c1153684e982b6ea7a43224c0509b8b0121983b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 06:37:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed0d5303b12a407e9121b5a4a38ab2b3 : SUCCESS in 4m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8ca77fef4fb0452bb4bbd85de3743a55 : SUCCESS in 1h 03m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c51cde22f374475f8935b7e9533cf195 : SUCCESS in 1h 12m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8362784c21b64cd9a880e5fbc1a7ce7a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b577da423634a29957e5f63613c894e : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a398cb707281479786a21a35231354ee : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d67029e3f18d4a048d8a3a2deb572274 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b47b57bbf14e4363a84ec7d9294c616a : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d6efbd54d6a48e6b93391c018a53c38 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/486492737cb04b5f9b2cca5bc84559eb : SUCCESS in 8m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/685922217c3e4a769af4978248c173d9 : TIMED_OUT in 31m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/324ee30dd6f64665819ac012a0ba3c73 : SUCCESS in 57m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac111df8a2904bdfba929729d77f6590 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4782bfb440cd5959b73d602e934e3086b95ef3e1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-06 06:39:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2256601674c24cbcd26607a5e3a4c75756b18715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 08:11:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/38d8bbe7e1074f93bba179c269c9e9b3 : SUCCESS in 3m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/afa1ba74ce554878bbe59b81d85cd626 : SUCCESS in 58m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/207dcec8a5aa45a8aba06c3144224107 : SUCCESS in 1h 13m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/877dd25060b0494c97e49bdcdfcc68ee : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d918eaefac847a992c4cb93b450b0f5 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4378a79f53a14fad91598d2f526066c5 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c3a64e18b364d1eae63014c5d2d2a53 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/589ff7e88e73459985dbc1b031aab874 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2270d21304d4fd0a8209db658b22cca : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2185954ee5be418d8f43c2839d721032 : SUCCESS in 8m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9f6d8b7e5c6f483ebfb8d378d329c45d : SUCCESS in 10m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/71cbeba15a2c46fca14776c2014d0fff : SUCCESS in 1h 01m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/265a811c86804c72968f59c34594fc57 : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d97b1cc33cec82fffab2fff03da48694303b2bbe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-06 08:12:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85122c2e1706ba336d0777065f4e1bc86a50d3fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 08:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b8864ce542569ce25eb53968df9d6fc843f08c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:32:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f9131a737d5b4bd8bd7ded01555c98bd : SUCCESS in 3m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b291906b2f64f7cb5de9cb0a44cd4f2 : SUCCESS in 56m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/546b740ad1bf491dbff183336e6723e7 : SUCCESS in 1h 18m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1500b9dd45634fbd88029bcfd30ccd6a : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0028f5c31a7643a9b803ae10d47f2255 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95de5ac183a547528ce31efc040f2cb8 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c6ade7a85a24e7caf2c8957f46b7085 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45a2acda2b64fd38bd3a30868a464b6 : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4eabcf364ec419aa3c0ce8304a848ca : SUCCESS in 9m 13s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f04960d557a64b3294782c5524918c5e : SUCCESS in 8m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5e9e34702c549f99c86dc34cb1bd3e2 : SUCCESS in 1h 15m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6d881a35cd1f7a94409ab1e13b62af5a6266c9c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:32:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"49f73b375d81e35706d1d2ca9116e92410bc8c3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:34:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85279e444f804e36885d7afe0b1a59b1 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7ddabb15f3a56570b59259a29c0e943805af007","revisions":{"c7ddabb15f3a56570b59259a29c0e943805af007":{"kind":"REWORK","_number":1,"created":"2020-03-06 05:24:00.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/81/711581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/711581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/711581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/711581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/711581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/711581/1"}}},"commit":{"parents":[{"commit":"5569caa5c185b82c09b37f6476e12c0ba1397180","subject":"Merge \"Amend tempest-irrelevant-files to avoid false positives\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5569caa5c185b82c09b37f6476e12c0ba1397180"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-03-06 05:22:52.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-03-06 05:23:58.000000000","tz":-360},"subject":"Remove policy.json from setup.cfg","message":"Remove policy.json from setup.cfg\n\npython setup.py install fails due to https://review.opendev.org/#/c/694387\nunable to copy missing file.\n\nChange-Id: Ib02387c4be70581df64cf5009750b4f30ae30159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7ddabb15f3a56570b59259a29c0e943805af007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7ddabb15f3a56570b59259a29c0e943805af007"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
