)]}'
{"id":"openstack%2Fglance~713720","triplet_id":"openstack%2Fglance~stable%2Ftrain~Ifd25ed15211ebc864f3a61558834bb1dae16a062","project":"openstack/glance","branch":"stable/train","topic":"fix_reload","hashtags":[],"change_id":"Ifd25ed15211ebc864f3a61558834bb1dae16a062","subject":"Fix reloading under PY3","status":"MERGED","created":"2020-03-18 19:23:10.000000000","updated":"2020-07-23 09:12:16.000000000","submitted":"2020-07-23 09:09:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713720-1595495400132-d04e827d","meta_rev_id":"573702ea71b9b4541dd53e2c8b7ccd48defeae33","_number":713720,"virtual_id_number":713720,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-22 22:33:40.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 09:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-23 07:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-23 07:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-23 09:12:16.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 22:33:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-23 07:45:11.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-23 09:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c0091d06427800436b8b613f7a3f3f55b92127e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-03-18 19:23:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a389ccd99bc8fcbd9494d0172792a24746d3725a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-18 19:32:41.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nJust a comment about the commit message, but I think we should change the utils.py changes to try to keep them as close to the changes on master as possible.","accounts_in_message":[],"_revision_number":1},{"id":"45957bafbd56ceca12e169a59a0053eda2958531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 21:17:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b6f1c73bfa1a45e98ceb2724d20a8e36 : SUCCESS in 1h 50m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dbf0ed71a9fa486bad4f4f3425972dc8 : SUCCESS in 1h 00m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c25b21151f741e8b14bbfdc88614628 : SUCCESS in 1h 05m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dee3f7a25fd41a3beb52278d21eee71 : SUCCESS in 1h 34m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb24ebfe2fb94ce3a3d93a8811013a8a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c567d07e116421ba5d677dd045594bc : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5abe9e510f79410eaa6d8be40c36a34b : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9af563e5386443c8fcddd22873e7f52 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7cecfc2bf9c430481a2e8506d8c1c1b : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd85e60466d24373ab0b2a0cf8bd13e8 : SUCCESS in 6m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/07e7c196391641b78bc94754a26319dd : SUCCESS in 9m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/baad21fcc24145b68ebc384ddf4514b2 : SUCCESS in 12m 41s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/7b4760edc7f64b628d0a960679398b10 : FAILURE in 4m 49s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f3b3ce9041264b898fc1ba7cc6cb1736 : SUCCESS in 5m 08s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77ce03f0a0da44efbb90bf8845cc4b7f : SUCCESS in 1h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88ed40386d2d259a94c0de39b31dab996c8e7f4c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-26 03:37:31.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good to me as the fix is proposed directly to stable/train branch.","accounts_in_message":[],"_revision_number":1},{"id":"15a01a20c17ed0f7c2dde17cf1e8b50d6947c064","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-22 14:32:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6e7dd7ddcec5fe24349770d090002756da504de","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 14:35:07.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"30b91df1225f2b41b9c0f7fc36f72d0c7a9dce9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:30:46.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fbdcee2a6bee4d8ab4cba8408bda94d8 : SUCCESS in 1h 40m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4984d83efcea4b83a1800708343c4bf8 : SUCCESS in 1h 02m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c43b04b45ea342fd8100f675020ca22c : SUCCESS in 1h 02m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e8b0cc8732a4b3bb0017e002d97d1c8 : SUCCESS in 1h 23m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a696e7af532847b582d7f21c3313bc61 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa62dd86ba5434da6c19a2a61d673e5 : FAILURE in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48ccc7f2e6fc4701aacea59c270526de : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7425a56f39b48d18b278fa843ea1792 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4399908d07149688d7a6733622d3137 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a688d494bc14466e9e55610bb15e6648 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2139f92dd15e4670b6e63abdb79b2915 : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bec6e68412cb4d77bd81ab7e6cd7ae96 : SUCCESS in 10m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/04c8bfe0a4814fb4b84fd0333eb3da1a : SUCCESS in 10m 23s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/b1bc77b738e342d69f75e445e4960f71 : FAILURE in 5m 31s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3e17318da5da4a8195e9a082c0197d5a : SUCCESS in 5m 45s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b67728f145149149bacada98adf1eca : SUCCESS in 1h 01m 33s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/189f0f60578846aebd46e907c0944490 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b6d55bd7ad71311f247bfce5e57a4e9301cc2f77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 18:51:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9187d4bdbfa8a3935d0ae4714092dd591824664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:03:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/360a17d438e842a9a040ea312d697089 : SUCCESS in 1h 25m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c671ded1cbad42dbb5febd08476c72b6 : SUCCESS in 56m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/75caf6606c33467f86555c5348794130 : SUCCESS in 1h 04m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d8b39f0c7ae46a88661c30baebaafe4 : TIMED_OUT in 2h 06m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb1dfd489d44e0ca9e9389940c6979f : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb85b5e0ca38482c9198ceeb9413e58d : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/289c2b96c159432a86cfa1daf8111fb7 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba979c36dba1416e8b8d7af9ef623939 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cf5da77ac3c4020974365fe887b9d8d : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa0ebb9752f426289e90fb117cc5b7e : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaf6c7c2b67043259578c7f4d88c81e0 : SUCCESS in 5m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b691ed05c321493bbedc84ce53a62766 : SUCCESS in 10m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/808a521ad3c04d058400c3db780bec32 : SUCCESS in 11m 50s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/cc72dac98d104558b3e851c2bb7a967d : FAILURE in 5m 36s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ebe92b7af55d4776a7710c364d1769c0 : SUCCESS in 5m 10s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ea8c63e41de4b8297d9446fc8f0a584 : SUCCESS in 1h 19m 33s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/47c9cbbc938c4508b03d387d0abc1869 : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ce7ce608512bba5209a0d7b468ff72b529491351","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 22:33:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"39a4b835ebc065825ee5c1fd79dac08c0ad63a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 23:56:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b32e0c7d0ea14d3fb2674caf7639e793 : SUCCESS in 1h 19m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9230deff30be46f68353646ea2a62f04 : SUCCESS in 1h 20m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/fb1f5ea2467a4461ab9f4bbf708dcc5d : SUCCESS in 1h 02m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49415f3347be41378ba2996922c235a7 : SUCCESS in 1h 10m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d56ea16c49f4ee1acb0cfb6a715acc7 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81195e06466c4d1ca49879aa64c37bd0 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f9e57fffb6841e6a32c6a1d70bacdc4 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0fc8b0c15ba4fd0a28f3f53d367877e : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d2b1408ae534bb0a9b6bb10ac6e2018 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c41659116f7948faa2fcc040e2ce6f83 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b62e8c4531149219ccf7292a9ec996e : SUCCESS in 4m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/124bb77a15cd4cd1b0373f23a920cdf0 : SUCCESS in 8m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/560bfc3c9c054cd8b49dcd9d6911566d : SUCCESS in 11m 02s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/040e571b58ab4ac9a57f88f1f57c3025 : FAILURE in 4m 28s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0157479abf864488acd52a98a79275d3 : SUCCESS in 5m 31s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15de2f13aa7a4927914346aadd994b67 : SUCCESS in 1h 08m 07s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4f61bc8def3149e58b2cd74a1dc3f360 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":3},{"id":"653254fadc26164cd0ea37bd7ddaac01fef2d584","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-23 07:45:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI am carry forwarding Sean\u0027s +2 on patch 2 as he just fixed pep8 error and submitted new PS.\n\nThank you Erno and Sean!!!","accounts_in_message":[],"_revision_number":3},{"id":"4dbc4e4f162abf0ad33f6270d3458135ae379574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 07:45:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"46ebc36ca5bb2bc6e18d5c5a2f2e37f015bffb9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 09:09:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb1ed6fe5a7f4939b5c28e13c73bb1d2 : SUCCESS in 1h 21m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5bc25a8a96a64b1d89146c9d9c1aaf20 : SUCCESS in 1h 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d383159cca4b490a81d542a22de815a4 : SUCCESS in 1h 16m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c15a030c4664bbdb17a9060a7bc5436 : SUCCESS in 1h 15m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/458ad7b5bb6148038ac7fcf109ae76ba : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11d3b4049024ceab3138a8a7ffa84a0 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd2b478caf614817bb7da3813225807c : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca725c462f524aa5900a252ba40b6c5e : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34ba7db4bdf1433eb46e15d1ecfcf0a0 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8611ca110f4335ae2a26dd2e6b2f31 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b8f13bf60e4addaafb6dff0391cc7b : SUCCESS in 5m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aa86639ee08f489395013df3f252d694 : SUCCESS in 10m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1b1670b1869844198a6edd5acff2e466 : SUCCESS in 13m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/67a114092ed5424e93bfc81b4235dff5 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":3},{"id":"132820a9ae4aa2a3bf848fc8ddb2104173f5f766","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 09:10:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"847556ab5581b26d11be89cd0e226d0c4f4e53ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 09:12:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd635296d3bf4b17b0b733cd4585d284 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d821ba6b3b404354a959d57dd5baecd5 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f3cbe90b3eab1f1952798937f177f23c4ef34e17","revisions":{"1afeb8cbc1cdf323ef600616d70cd7c3f0a9a1c0":{"kind":"REWORK","_number":1,"created":"2020-03-18 19:23:10.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/20/713720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/713720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/713720/1"}}},"commit":{"parents":[{"commit":"3e16bc7d40922f4e6176486b5e7af42989a50774","subject":"Merge \"fix properties\u0027 missing underline for VirtCPUTopology\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e16bc7d40922f4e6176486b5e7af42989a50774"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-03-18 19:20:10.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-03-18 19:20:10.000000000","tz":0},"subject":"Fix reloading under PY3","message":"Fix reloading under PY3\n\nThis is not direct backport of I361c9d0b53a67b1cc60f0f3012e22accdcc46055\nbut rather adaptation of it to facilitate SSL/TLS termination under\npy27.\n\nCo-Authored-By: Khuong Luu \u003clcamkhuong@gmail.com\u003e\n\nChange-Id: Ifd25ed15211ebc864f3a61558834bb1dae16a062\nCloses-Bug: #1855708\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1afeb8cbc1cdf323ef600616d70cd7c3f0a9a1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1afeb8cbc1cdf323ef600616d70cd7c3f0a9a1c0"}]},"branch":"refs/heads/stable/train"},"5d2719f164ba841d391bf66f73a14220478bb5c5":{"kind":"REWORK","_number":2,"created":"2020-07-22 14:32:53.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/20/713720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/713720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/713720/2"}}},"commit":{"parents":[{"commit":"3e16bc7d40922f4e6176486b5e7af42989a50774","subject":"Merge \"fix properties\u0027 missing underline for VirtCPUTopology\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e16bc7d40922f4e6176486b5e7af42989a50774"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-03-18 19:20:10.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-22 14:32:37.000000000","tz":60},"subject":"Fix reloading under PY3","message":"Fix reloading under PY3\n\nThis is not direct backport of Ussuri fix\nI361c9d0b53a67b1cc60f0f3012e22accdcc46055 but rather\nadaptation of it to facilitate SSL/TLS termination under\npy27.\n\nCo-Authored-By: Khuong Luu \u003clcamkhuong@gmail.com\u003e\n\nChange-Id: Ifd25ed15211ebc864f3a61558834bb1dae16a062\nCloses-Bug: #1855708\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d2719f164ba841d391bf66f73a14220478bb5c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d2719f164ba841d391bf66f73a14220478bb5c5"}]},"branch":"refs/heads/stable/train"},"f3cbe90b3eab1f1952798937f177f23c4ef34e17":{"kind":"REWORK","_number":3,"created":"2020-07-22 18:51:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/20/713720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/713720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/713720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/713720/3"}}},"commit":{"parents":[{"commit":"3e16bc7d40922f4e6176486b5e7af42989a50774","subject":"Merge \"fix properties\u0027 missing underline for VirtCPUTopology\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e16bc7d40922f4e6176486b5e7af42989a50774"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-03-18 19:20:10.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-22 18:51:29.000000000","tz":-300},"subject":"Fix reloading under PY3","message":"Fix reloading under PY3\n\nThis is not direct backport of Ussuri fix\nI361c9d0b53a67b1cc60f0f3012e22accdcc46055 but rather\nadaptation of it to facilitate SSL/TLS termination under\npy27.\n\nCo-Authored-By: Khuong Luu \u003clcamkhuong@gmail.com\u003e\n\nChange-Id: Ifd25ed15211ebc864f3a61558834bb1dae16a062\nCloses-Bug: #1855708\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3cbe90b3eab1f1952798937f177f23c4ef34e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3cbe90b3eab1f1952798937f177f23c4ef34e17"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
