)]}'
{"id":"openstack%2Fglance~715449","triplet_id":"openstack%2Fglance~master~I9d030eb450f2c7ae74c25b7564a01b8785503e5e","project":"openstack/glance","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I9d030eb450f2c7ae74c25b7564a01b8785503e5e","subject":"Cleanup old cruft","status":"MERGED","created":"2020-03-27 15:41:17.000000000","updated":"2020-03-30 06:40:08.000000000","submitted":"2020-03-30 06:36:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715449-1585550203722-f250c9cc","meta_rev_id":"8d010c602fa91b18f8d57ae21ae67248b68398f0","_number":715449,"virtual_id_number":715449,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-28 07:18:28.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 06:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-28 16:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-30 05:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-30 05:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-28 07:18:28.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-28 16:45:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-30 05:22:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-03-30 06:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a44bb4908f4b37d791495601ea8b9dbc64f0c90","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 15:41:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"264dd79417f6dab5ba6dc0f93f0161f11eafd5a5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-27 15:41:37.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"2e6c19ed94621086d0cea677ef400da9fd31aa22","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-27 15:45:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24a007ba768c4f898e71bc80ce517146a5ca5d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 17:43:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/889ac829c4e7477eb6341216b9d316aa : SUCCESS in 3m 43s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/be01857144b5487b9e5b0f7bc772533e : SUCCESS in 6m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d190d2fe6b9e4e7bae9ef34ba6f3dda6 : SUCCESS in 4m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ec5be381455e4d388975725b633562a8 : SUCCESS in 59m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/756f8074501246178e02fff4bb5c006b : SUCCESS in 1h 25m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd931823c5d04d538a1597a810305a6e : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7297ad3875443c8a833f43273bd9c50 : FAILURE in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6d78c7b42a84a6dbc2759c8a2fd04e4 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53cbefb1e8bd419b873f4abe47b9f6fd : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98a490695bd24c8ebf90afaed6b163db : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f526a7236d484b9009d00f20fb2cdf : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215c8673eb614ec2a9fcced1aecc41ba : SUCCESS in 11m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f02a6abc36b0427485f84366461e6b64 : SUCCESS in 12m 05s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/97fac5b9f7184df0829711a505d06ca8 : SUCCESS in 10m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/46d88e444eab4426bfec9c7cb1c9eb19 : SUCCESS in 6m 19s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8be1af84c0134cbc9ed58b3e4126fd64 : SUCCESS in 1h 05m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39098ef7e0434f98920bf56380d1a32b : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a9568a37bc113387c8d06a3a2d55adde08469a96","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-27 18:56:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c7e0a14ba1dbdeaebae7f7b3a7147652997f8b1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-27 18:56:36.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"06b28ebc5e1aa79eee59dd5e3062abfa91082870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 20:11:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fca55af4872b43f6a501e425e12526ff : SUCCESS in 5m 18s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7d93ea3183f14989b148e0dafab0796b : SUCCESS in 4m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24a200a4741241d2b66367f8cecc6cb0 : FAILURE in 3m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26d795cb7e9748ceb0115f45b4de87df : SUCCESS in 1h 02m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5b885ef817a477485260c9e15eb76bd : SUCCESS in 1h 09m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b14cdefc861c4f44b71827dbbf88c20d : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c69b17fde6486b8670d4d1d7be2d44 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44e7a217888f427c8a129cf1f308b12f : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81d9cbe4199c4b06b292af80f39920c4 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/910e652a37f44da98439c5aae35182e9 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02f8dfa29e3436b860ffd6bc060c5f8 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05cde567f737429eb8358ee1f0365b26 : SUCCESS in 9m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df7f86ae022b4976b059a94dde1ca1e8 : SUCCESS in 8m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/df10c4a5b5b54f5d8598529f7f7d7bcc : SUCCESS in 7m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bc1f1ab69aae42d8a28d7ec5bbae6266 : SUCCESS in 4m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c361cf00497d4a1592de919b6a06fa4d : SUCCESS in 1h 00m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4e159bc78384afe9e15093d6e332e1d : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fcd0b64a5de396455907ed8f0c122d0a1ca02d57","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-28 07:18:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2df1de280fc2ae57a1db1e6d9af89105e0e10b39","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-28 07:18:28.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"6debfa2a44c124c0849008fc7d9d015a416b843e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 08:33:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a41e745fbbbd41239490305ae16c7e45 : SUCCESS in 3m 31s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e56218935c934741b01e7fc29b70555d : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/999c34ba5aeb416b9ef9f50dbf92ae46 : SUCCESS in 3m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8a460f17f4e748d59860a72897ad2437 : SUCCESS in 1h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b9d3e333b4942e48179b0face188255 : SUCCESS in 1h 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4abccb6b33a9407f892d9fe38e119f3b : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d3bd092c74415dbdf7696b0f0bd906 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9775c5740ba0499883afb5d68be6b488 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4d4f482d53c4aa8bcdb9337e219d931 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4a5a0335ebc4b16b0014894e203e916 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20843a07b5214ad2b86f7309c552fcbf : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d795d9b8634e35a111bd32d69a12f1 : SUCCESS in 9m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/392ca19b250c49c7909f74d97d892f16 : SUCCESS in 6m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/446639c5f2314161b528f14cbbfb7e8f : SUCCESS in 7m 43s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f62ad155616c44c8b9898e798b42caea : SUCCESS in 4m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a29ff3cbf19644a2921919a7f2cccbb2 : SUCCESS in 1h 03m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bcec96cf6f8441ab1094734f7c6b701 : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":3},{"id":"3f8dff0f2ebb9a6656efbe7b1b286c93a469f254","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-28 16:45:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3ca8fd8bd03d6e305f774c475d50fb71a655bf2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-30 05:22:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you!!!","accounts_in_message":[],"_revision_number":3},{"id":"0dd403faaeeeb5b558bc8cd0e67051529cccf2af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 05:22:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9872aa471dd82674ad34c118a78acf22c2fef48b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 06:36:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"eccab6f25aedec1ffa33a8885b726db90e37c6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 06:36:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ff4e8c8705fd4870bbac412202282f2c : SUCCESS in 2m 55s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5393190648a840c8b1013bd72e1a6153 : SUCCESS in 4m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/db0ff2546b8b4076880946f9af7d2622 : SUCCESS in 2m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1f20e9de0dee431b8666c2aded39ef56 : SUCCESS in 57m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed6240774331448b85b21cace65ee298 : SUCCESS in 1h 13m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5a009be60754c69b44e41d9e51dba89 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1b255b47664a8ab133855a7a85b947 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23732bb84b534142824528d5c8b46007 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6fbd62672214209849b823a4e1dfccc : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01c92bd1ebd448c806a4c0185cebb6c : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be8327fde551492d8772af9f3c991914 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2b15e4a38dd46e19228aa223b960dac : SUCCESS in 6m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3cf98bc3a8ac4c21a4de4113346f0525 : SUCCESS in 7m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2944d148e5ba4c6bbb150a6eab9053ce : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":3},{"id":"195dba3ea96f64b39ad35c03d30599663160698b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 06:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0276315c7142406284187c17677c09eb : SUCCESS in 2m 31s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7e8a43d06a4cfcabd628071bf7520f : SUCCESS in 2m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98490c9dcd4c45e5b3240513600bcf67 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"176d24de94c5e2007017dafb7adcb75911832901","revisions":{"f0659edf3c3362a87974e8058e469e3ad0cc5013":{"kind":"REWORK","_number":1,"created":"2020-03-27 15:41:17.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/715449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/715449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/715449/1"}}},"commit":{"parents":[{"commit":"2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca","subject":"Merge \"Add decompression import plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 15:34:22.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 15:40:54.000000000","tz":60},"subject":"Cleanup old cruft","message":"Cleanup old cruft\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Update requirements: Remove sphinx, it\u0027s not needed for testing\n- Update doc/requirements: Remove python 2.7 support\n- tox.ini: Remove testing of po files, the infra scripts do this since\n  a long time\n- Update conf.py, no need to import openstackdocstheme anymore.\n\nChange-Id: I9d030eb450f2c7ae74c25b7564a01b8785503e5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0659edf3c3362a87974e8058e469e3ad0cc5013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0659edf3c3362a87974e8058e469e3ad0cc5013"}]},"branch":"refs/heads/master"},"206d3ce4b5bd2af4d30afad3b3957c9ccf584c08":{"kind":"REWORK","_number":2,"created":"2020-03-27 18:56:15.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/715449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/715449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/715449/2"}}},"commit":{"parents":[{"commit":"2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca","subject":"Merge \"Add decompression import plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 15:34:22.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 18:55:39.000000000","tz":60},"subject":"Cleanup old cruft","message":"Cleanup old cruft\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Update requirements: Remove sphinx, it\u0027s not needed for testing, add\n  pygments which is really needed.\n- Update doc/requirements: Remove python 2.7 support\n- tox.ini: Remove testing of po files, the infra scripts do this since\n  a long time\n- Update conf.py, no need to import openstackdocstheme anymore.\n\nChange-Id: I9d030eb450f2c7ae74c25b7564a01b8785503e5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/206d3ce4b5bd2af4d30afad3b3957c9ccf584c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/206d3ce4b5bd2af4d30afad3b3957c9ccf584c08"}]},"branch":"refs/heads/master"},"176d24de94c5e2007017dafb7adcb75911832901":{"kind":"REWORK","_number":3,"created":"2020-03-28 07:18:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/49/715449/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/49/715449/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/49/715449/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/49/715449/3"}}},"commit":{"parents":[{"commit":"2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca","subject":"Merge \"Add decompression import plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d80135f9b459e3c194c0d9e5bf62d4cef2f77ca"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-27 15:34:22.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-28 07:18:01.000000000","tz":60},"subject":"Cleanup old cruft","message":"Cleanup old cruft\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Update requirements: Remove sphinx, it\u0027s not needed for testing, add\n  pygments which is really needed.\n- Update doc/requirements: Remove python 2.7 support\n- tox.ini: Remove testing of po files, the infra scripts do this since\n  a long time\n- Update conf.py, no need to import openstackdocstheme anymore.\n\nChange-Id: I9d030eb450f2c7ae74c25b7564a01b8785503e5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/176d24de94c5e2007017dafb7adcb75911832901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/176d24de94c5e2007017dafb7adcb75911832901"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
