)]}'
{"id":"openstack%2Fglance~716058","triplet_id":"openstack%2Fglance~master~I95a8d8cf02a0cb923418c0b5655442b8d7bc6b08","project":"openstack/glance","branch":"master","topic":"bug/1863021","hashtags":[],"change_id":"I95a8d8cf02a0cb923418c0b5655442b8d7bc6b08","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-03-30 19:14:37.000000000","updated":"2020-03-31 12:57:12.000000000","submitted":"2020-03-31 12:53:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716058-1585659229266-264390e0","meta_rev_id":"eda8da39cdf1b0b228a38908837b1411cdac9587","_number":716058,"virtual_id_number":716058,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-31 04:49:42.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 12:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-31 10:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-03-31 06:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-31 10:57:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-03-31 06:28:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 04:49:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-31 06:28:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-03-31 10:57:00.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-03-31 12:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a3fc86e0b2cfacda9a40577bac69576744e68cc","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-03-30 19:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"587dbb84d1818468cacf0b58d2700ca0a2870e45","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-30 19:15:33.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"cee4bca8f8955a963b0fe2d6c10110761872b3b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 21:38:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4a335b01f134483cbda0267a4c2b52ab : SUCCESS in 1h 21m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3812d16f8b84a05b501b108db8f98ed : SUCCESS in 1h 32m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/274ccab4ff7c409a80a6346ef27cc920 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16b7542e1224d609fa0e4a7bc9bc4a6 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2da4ef0a5e4ffebe521847249a2e0b : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68c8ad1a7a12417788ac0f7c7fdd18d3 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e7b4ddb3743482b95216ce099106350 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e65715a392e4699a4c66ecbec469919 : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dba28bc05f349ac8b531415a1297c59 : TIMED_OUT in 31m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/db3ad36952d7424e8347dd6aae268ef1 : SUCCESS in 12m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1f1f5d3de4324baaa5ba8440faa3a02a : SUCCESS in 4m 57s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2c7d26e5350143f3a5d6a98a44a6569e : SUCCESS in 1h 08m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/264be90212bb4d13991b8e099ccf90c4 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"52dcce90183b9bcc28228b3f97e9fecb78cc7b75","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-31 04:48:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df15c6d4e2a79af35435145de9dd9a136630a02f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-31 04:49:42.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"6bed6dd195cea597d966d4aa635cc057ca7aab97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 06:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b1698bcf476845dfaafa8bf104d4757d : SUCCESS in 1h 08m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5247e15418147a7aa672a5d393fe029 : SUCCESS in 1h 34m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf15dfa0c91548c1a8e9a4d3fefdfd61 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd955a94e80f4defaa5790b7ee6a6d0a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d4f2ad8349e405ca737a95f5fcaa034 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee150e7f57d64a32af302b38c874f1d0 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcb19a2085464f94bba903a4902c6a9f : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e4fd31455145cfb9103ff555a7fd48 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49f36ea8439c4261a49754f7fdc73f1c : SUCCESS in 6m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5b8166b8019f4e5eb2a956e29112e9f4 : SUCCESS in 7m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/59c58e94413f4907b53f0372a3974906 : SUCCESS in 4m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f7bcfc2a0068464b8ec091af4526858f : SUCCESS in 1h 08m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2592b9bfaf945939e884de14d70353b : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":1},{"id":"961e1be19a4b013d3b0b42f53c2f76906a1c7dd1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-31 06:28:43.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"a1d46f0ff3b98695195e0f6f5f42883737b65a46","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-03-31 10:57:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGreat stuff, thanks Corey!!!","accounts_in_message":[],"_revision_number":1},{"id":"8578a9718e1ce0d205e8205c758b09038634b678","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 10:57:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54df592cf1b380d353b1ce9be08834483b222421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 12:53:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"76c2c15744ba8e97a8e4ae01754b3b68c08d3486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 12:53:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b0e5d4c70f0b486984de74601ed1ec0b : SUCCESS in 55m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d60a0a812cc24742b196dc0a6fa5a550 : SUCCESS in 1h 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e280ba39dfa49a7b3f6431db2209373 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80a10300206a4b60b19edca60a54cff8 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e48bf8ce8494fb9ac2670f8e3628cdd : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f76e4d59b6fd4f5db873013e82873add : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41f474a07d5417fac5d88b92207b04e : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ceb2eae4809a45e2a5623d26f23cbb4f : SUCCESS in 7m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b1dfdfb2d6f44d38a731b009a47d071f : SUCCESS in 11m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fbc9851a5944d18be6f60dbdc75b931 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a61082f949edbac1dc5ce4e8b712b23628924eee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 12:57:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088de6dbe233498d9b53e9c275a9086f : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90146b62c765ad1b8be1ffec1799cba9f3994c2d","revisions":{"90146b62c765ad1b8be1ffec1799cba9f3994c2d":{"kind":"REWORK","_number":1,"created":"2020-03-30 19:14:37.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/58/716058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/58/716058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/716058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/716058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/58/716058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/58/716058/1"}}},"commit":{"parents":[{"commit":"2143060833a7afaa095d43ed6d934ff70be5417c","subject":"Merge \"Raise hacking to latest 2.0.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2143060833a7afaa095d43ed6d934ff70be5417c"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-03-30 19:14:15.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-03-30 19:14:36.000000000","tz":-240},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I95a8d8cf02a0cb923418c0b5655442b8d7bc6b08\nCloses-Bug: #1863021\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90146b62c765ad1b8be1ffec1799cba9f3994c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90146b62c765ad1b8be1ffec1799cba9f3994c2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
