)]}'
{"id":"openstack%2Fglance~716078","triplet_id":"openstack%2Fglance~master~I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587","project":"openstack/glance","branch":"master","topic":"M3/config-refresh","hashtags":[],"change_id":"I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587","subject":"Deprecate admin_role","status":"MERGED","created":"2020-03-30 20:55:43.000000000","updated":"2020-04-11 19:47:39.000000000","submitted":"2020-04-11 19:46:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716078-1586634363576-76a6aaaf","meta_rev_id":"3362fd41d8311ffca490fc1e9ce0e7b8d7156ded","_number":716078,"virtual_id_number":716078,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-10 16:04:48.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-11 19:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-10 14:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-04-10 14:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-04-11 18:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-04-10 14:16:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 12:02:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-10 16:04:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-11 18:25:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-04-11 19:46:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c0e69c4676fe93b43186372d5e02e4ec1a42198","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-30 20:55:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d144c6c9911369c1f317ca912857d77fbd11ce0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-30 20:56:04.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"c9b8116337ae12e15c0bb9da95c92cef36f4c992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 22:27:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/daa5ca4bde3546d49d52a159ee58d276 : SUCCESS in 1h 02m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8743a0d5a60e443f950393cba1166f36 : SUCCESS in 1h 25m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0af54f04605f4b5e94c047a8c4f8155e : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/935b259f50884441a1ff7c64c7639225 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/428ea5b0b07a4c8480cbaa6f4eb2de31 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/542eca1c7ccb4aec9a57967d16009013 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/304f16729dad4ef38dd1807827200f53 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63183d21f9fb413288e5dde565703967 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c1ced99678c45fc8a2e7e66ee71a9c8 : SUCCESS in 9m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29b8050abd0a4e229d6a0b9d7ad17eb8 : SUCCESS in 7m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/463994a90ecb4aab9c1dfedef96423ec : SUCCESS in 8m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/af0b71e4c66747ecb141affd2c68c621 : SUCCESS in 4m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1952e9d0be954d6fb469767360a46cda : SUCCESS in 1h 01m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1681bb5b4f104a028c8d42eeed5ca437 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ac8728d23d6f174b36946e2f243d9614d9568a1d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-31 04:44:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLooks good, just minor nit, which needs to be removed.","accounts_in_message":[],"_revision_number":1},{"id":"6dd6f0c0de824181144112989e423d71d3cf9ca0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-31 12:07:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15a214e720b671523c735e718e4e93c3004415bf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-31 13:01:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e9d170544c77cc20fa6fcf3300949c7f833e2b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:47:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ea873a5addc4b48b3c1bbab73639060 : SUCCESS in 1h 03m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8356c22cf49143a699a7b710642dd3d2 : SUCCESS in 1h 23m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e25544dd504662b4ae82fa8756e5d4 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d1317fcb494d83a5a5eb866be2e94d : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8917de8662d34569b80bc233152d7746 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84730252fab240018fb7196a4c8af798 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f91c7f56e64e2aa417cae381864dea : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ddd53c6e8e443aa748e57327a88a78 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08dd3389da68468bbfc1b5c5831b89c6 : SUCCESS in 11m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/817103b982f84df8816541ab2a3e7e2b : SUCCESS in 9m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/05e3312e2b8340e7b7176fc2e88f9fb9 : SUCCESS in 10m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8c815912d2234927a380ee7a18e4bc7c : SUCCESS in 7m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e5749d9ea7134b7f84784a6d1977717c : SUCCESS in 1h 15m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f351f299f73a43249e22868d9f1a2a16 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8fbf760254108044ee36a6b80590df1268c2f087","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-31 15:37:31.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nThank you Brian!!","accounts_in_message":[],"_revision_number":2},{"id":"e39ea71126f9cfd029f39878a7e547dad4a59457","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-31 19:00:53.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"f66021e551ab65cd961c9ee4e2f3dc113da12745","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-02 15:56:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d9e78169565e7f544c6f189d86589335505918f5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-04-09 16:27:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cca2c0c76766321568e1f63641e3d662686d783","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-09 16:28:07.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"e694ae7d4b7637c71e2928c8a8aae539444df213","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-04-09 17:21:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"179f90e2dee26f4cdea189c098f8de4a0474f894","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-09 17:22:09.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"6178017710ca54f91dc3552e8292b24a3b884b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 21:07:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/98694483b9574bcc86a2f47ea44db33e : SUCCESS in 1h 29m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/12618bb74be14312815ea0223afe8d06 : SUCCESS in 1h 31m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c69c1c01d8f49a3b80035543f11186b : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d237a7bd1922454cb7c8fe3b753e63d1 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eda4001364f41ef8b54b30df7470f0c : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/684fdf5f3d6743a6bbc286d5f55f5414 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/454e733206794911bc41107caa550165 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ba081d57284fa0bddf1805b448b9a7 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc35f6ddec174415a8517d1e05328d0e : SUCCESS in 10m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb468265f4fe4c4a8c761ab0cae35862 : TIMED_OUT in 31m 25s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d605c511dc49400fa683800b3471bee9 : SUCCESS in 12m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/63b9a54b25db4f3a8b9be40d2136df26 : SUCCESS in 7m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dee859d4d7cd43008d95d9c7c6bfea92 : SUCCESS in 1h 03m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfcba7318c234c16b020a9650b2f3b65 : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":4},{"id":"c35562eaa2133f7b964f433c6ef94bd5895a52a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-09 21:41:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"832f07889b3a60a3a27b6c2ce115674cf43e75d8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-09 21:42:10.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"ec2efbcfcae1de26e6d2f95d80009d9a19946612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 23:37:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e670d474aabe4450ad91c2a49e37cf2c : SUCCESS in 1h 12m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d054d3bd79e9449bb54b10e74d5e15e7 : SUCCESS in 1h 26m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9e73824ac7f4f08b6503de6b33f7612 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f520f8362ca46269c8bb3d76261a07e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8bd43ca7084d2f9e7d71daae32b26e : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ab2cd7e439841d0936f0a04a44bfd47 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f69536d23664961b3d57efd19435aba : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01445c22ed043319d45736405e9bc51 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7df1207b3fb4ee188cc65351e1707c6 : SUCCESS in 11m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f66518188e14cee8918202b6264024d : TIMED_OUT in 31m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/577621d2ff0944bea80f1e955f8be755 : SUCCESS in 7m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4572bf41615b4ed6971e91dfcf989b5b : SUCCESS in 6m 21s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0ab143bb904c4b779252d5e615a6e767 : SUCCESS in 1h 00m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3e14ba8ef54f2dbd850ba5d1904c92 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":4},{"id":"15ef8efe5e5a446d90a3ccdc70e927a858634011","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 04:26:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ca970e2476041c053200f3152c91eb3da087ef7b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-10 04:26:57.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"29df3a750d9df0b143818458d106a356f93a7382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 05:44:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/862e9513835a435f86000202de165c67 : SUCCESS in 1h 09m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd57249bfdde48eb9b12675956ac0801 : SUCCESS in 1h 14m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85ff518b2c1d419e98a2c7154a184e50 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8f6911d120d42479e6925eae764a4ce : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7931348e9644c5c9c4f419518396bc1 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32d9b42e485a4d70819a2d59a00ed91a : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36f62d1ca9d747f9ae0c838186b29031 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/509fdf4757f048d280afbd75bd290326 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9aec81b4a24ca0b28ef4d7caede120 : SUCCESS in 10m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38a6900447ec42bcbf0f7774d7dc22bb : SUCCESS in 6m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c2713f86c570418598ba733da538383d : SUCCESS in 8m 25s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3972d773a2124113a18dbdd1ca23c3f7 : SUCCESS in 4m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f05a5bbf3377464e988b8b24d4c22407 : FAILURE in 1h 01m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35593ccbfde04982a0993c44e1703b42 : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":4},{"id":"5e9f9e721e1f36eaac1597d134e30f6296f0bc73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 05:55:19.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"1de3ab824f3a9f7b4e5c13fb32d022cdadb0fe17","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 12:02:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bd50312a882d81d798cef0891e2c08641d92ea65","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 14:16:26.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6c51c528bf6a0bea8f29b277848f98555f7464a9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-10 14:16:52.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"b657321cecc0044d1bc01fb7675c2fe6075a8736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 15:39:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9aebae45be440a8a1d4432296139948 : SUCCESS in 1h 05m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db549472eaca45f7a381d7b2b99a5845 : SUCCESS in 1h 16m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3672dde03082497b99489113ea7c778e : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3099911ff5a742e6b77dd6c320d5f12f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d80bea67fc4439b1ac6ea5b48bd8b2 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43bfc8e8d70547f1851d3761414ae75a : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69acff0d1ebf4ff08fcade3779c03692 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a5d4bc0145408dada0cf49652eafde : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6efa7201173442a9d6dd57d2edaa307 : SUCCESS in 8m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/970ef4277cd545be9357a8af0a9830f7 : TIMED_OUT in 31m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4995aba23d574307b027e251ad3b2de5 : SUCCESS in 11m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/37565d297c70478c8f9103a2ad373869 : SUCCESS in 5m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/19e1fdd966914d83b6fd02e2bc7ea845 : SUCCESS in 1h 06m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93208f9c69bf4068b8d40ddd74e0f955 : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":5},{"id":"87f88f587861482afa4f022b17e7782f02995bec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 16:04:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fa142d376a638d43eee735ef58cc8a6c80333a7f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-10 16:04:48.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"f104c68b48a30abe8e8f4ecfb21d3459b4b53de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 17:25:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1b33bc33f6f74155b0543ce953f6dd97 : SUCCESS in 1h 12m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52c4e801de9344ea9e7b338d4ad82d3c : SUCCESS in 1h 20m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0894816c6c97498893f654242d1cfadc : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768e30bd7fef4916ae6653c3389cc7df : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f560c1ed020c43009846b6fbc22e4482 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5cf5cd90286456d899d69848d47db70 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e38cfcd6c1548cba8a9966750256940 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a215d731d1468989d7f349166a755d : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee3ab996d91d4182809900c937ca2ba3 : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a972e9d70a784c5dab6bb3bb42009eeb : SUCCESS in 14m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8230375444724ca4bfa750277276b963 : SUCCESS in 11m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/81482c0acacf4d738ecfc6b35ebf9bed : SUCCESS in 5m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1878851163274722a10d989352384676 : SUCCESS in 59m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68fc92df2e7c42bba79efa442fe2c97f : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":5},{"id":"3277011db3e65e1e037ea4c06c81efb0b8146008","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-11 18:25:29.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2e40fa0f0962aafc82e35c038ec9fea723315470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 18:25:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"749eb80529c752dffeb946eeb0f32010abe1d3b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 19:46:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c4203456695649bfb6dbd80ee2e53dba : SUCCESS in 1h 08m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b224ec07f6384d89812115b9658c75da : SUCCESS in 1h 19m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6371290e1d7d4c86975aac0bfc3dc365 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8727a15a843434bb3acadc4362561d9 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d46c0fcbf6b44e57aa0e0156c3821e81 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebd5a1595c364613a88c8d55f15991ab : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03653e89c4a245cbab230706d81b1f19 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec280ed4f1004c4abf12b47fd87471bd : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cd11dd29e734624a81151b13e189508 : SUCCESS in 6m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/70bb14534f4d407b837fe6e7a6651b57 : SUCCESS in 7m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3569d68ebe24dd290aa885a6f964878 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":5},{"id":"2f2b43d5792b8fe9312f460bdc294d14dd7a7ca9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 19:46:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"26917a0eaf3d968d217d37bd73a56914dc168fce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 19:47:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628d140fda60438fa990ad3816c551d6 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9efdc17a50da40b087fb6a06fc797371 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"28afc9fc6fc49532be6d5dd51eff8b61a9d1b4ca","revisions":{"8cf5038a5e82154738bdcfc0f4865f1bd7c87ece":{"kind":"REWORK","_number":1,"created":"2020-03-30 20:55:43.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/716078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/716078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/716078/1"}}},"commit":{"parents":[{"commit":"2143060833a7afaa095d43ed6d934ff70be5417c","subject":"Merge \"Raise hacking to latest 2.0.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2143060833a7afaa095d43ed6d934ff70be5417c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-30 20:49:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-30 20:51:39.000000000","tz":-240},"subject":"Deprecate admin_role","message":"Deprecate admin_role\n\nImplements: bp deprecate-admin-role\nDepends-on: https://review.opendev.org/#/c/714626/\nChange-Id: I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8cf5038a5e82154738bdcfc0f4865f1bd7c87ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8cf5038a5e82154738bdcfc0f4865f1bd7c87ece"}]},"branch":"refs/heads/master"},"065af79bdd59aa22e8877919811f4a46874c1b4a":{"kind":"REWORK","_number":2,"created":"2020-03-31 13:01:55.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/716078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/716078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/716078/2"}}},"commit":{"parents":[{"commit":"2143060833a7afaa095d43ed6d934ff70be5417c","subject":"Merge \"Raise hacking to latest 2.0.0 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2143060833a7afaa095d43ed6d934ff70be5417c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-30 20:49:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-31 13:01:46.000000000","tz":-240},"subject":"Deprecate admin_role","message":"Deprecate admin_role\n\nImplements: bp deprecate-admin-role\nDepends-on: https://review.opendev.org/#/c/714626/\nChange-Id: I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065af79bdd59aa22e8877919811f4a46874c1b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065af79bdd59aa22e8877919811f4a46874c1b4a"}]},"branch":"refs/heads/master"},"3cd3a905d102657e6f62ebb23a27cb830d33ee45":{"kind":"REWORK","_number":3,"created":"2020-04-09 16:27:46.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/716078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/716078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/716078/3"}}},"commit":{"parents":[{"commit":"a14bd83f3e55dba7f1134450747f932bfc6e44a8","subject":"Merge \"Do not decompress \u0027compressed\u0027 containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a14bd83f3e55dba7f1134450747f932bfc6e44a8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-30 20:49:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-04-09 16:27:38.000000000","tz":-240},"subject":"Deprecate admin_role","message":"Deprecate admin_role\n\nImplements: bp deprecate-admin-role\nDepends-on: https://review.opendev.org/#/c/714626/\nChange-Id: I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cd3a905d102657e6f62ebb23a27cb830d33ee45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3cd3a905d102657e6f62ebb23a27cb830d33ee45"}]},"branch":"refs/heads/master"},"2b2a736140d86ca0f284aa1bc697b644501a5b71":{"kind":"REWORK","_number":4,"created":"2020-04-09 17:21:47.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/716078/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/716078/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/716078/4"}}},"commit":{"parents":[{"commit":"a14bd83f3e55dba7f1134450747f932bfc6e44a8","subject":"Merge \"Do not decompress \u0027compressed\u0027 containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a14bd83f3e55dba7f1134450747f932bfc6e44a8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-30 20:49:58.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-04-09 17:21:29.000000000","tz":-240},"subject":"Deprecate admin_role","message":"Deprecate admin_role\n\nImplements: bp deprecate-admin-role\nChange-Id: I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b2a736140d86ca0f284aa1bc697b644501a5b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b2a736140d86ca0f284aa1bc697b644501a5b71"}]},"branch":"refs/heads/master"},"28afc9fc6fc49532be6d5dd51eff8b61a9d1b4ca":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-10 14:16:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/78/716078/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/716078/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/716078/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/716078/5"}}},"commit":{"parents":[{"commit":"d5a0ce18aa4dbc3e8d487326d509e213d1b92bb6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5a0ce18aa4dbc3e8d487326d509e213d1b92bb6"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-03-30 20:49:58.000000000","tz":-240},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-04-10 13:47:42.000000000","tz":0},"subject":"Deprecate admin_role","message":"Deprecate admin_role\n\nImplements: bp deprecate-admin-role\nChange-Id: I0f61f85a0aaa4f68e345fa08fbb6b039d3d32587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28afc9fc6fc49532be6d5dd51eff8b61a9d1b4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28afc9fc6fc49532be6d5dd51eff8b61a9d1b4ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
