)]}'
{"id":"openstack%2Fglance~717716","triplet_id":"openstack%2Fglance~stable%2Ftrain~If86d805ddc46944e3aa2785aa5d775fe15b8468c","project":"openstack/glance","branch":"stable/train","topic":"bug/1823703-stable/train","hashtags":[],"change_id":"If86d805ddc46944e3aa2785aa5d775fe15b8468c","subject":"Use application_url in API version document","status":"ABANDONED","created":"2020-04-06 11:34:52.000000000","updated":"2022-05-11 14:30:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa3aff56060bb230c876cfa1eef6f3d797ab9ff5","_number":717716,"virtual_id_number":717716,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-24 01:13:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-23 23:34:57.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-06 12:37:19.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-06 12:37:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-24 01:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"069451f1fae1155cad10c7dd1f4f0a9507496d79","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-04-06 11:34:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f05af7b407286f82aef676b3b6bed2717c2bb0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 13:10:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/03e83d7f66e942e98588ad257052e66b : FAILURE in 48m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5fb4af88e95d44e5ac6c344f6f0dac24 : SUCCESS in 1h 16m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6db249d1daf844aea08197e90c2a33b6 : SUCCESS in 1h 15m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12abeb8e0b694cf18821be16e87c723b : SUCCESS in 1h 33m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/538c1103022042a8a6913b28b9fc4f9a : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d591f8eec97f47c1af48cab2bed46de1 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cc58cd40ce84b4998460667f0cb1771 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf662c51843d44b2b643b540f23078ea : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff852f2289c84540901f6f5b9235b705 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fae023b03ac4f5599a43a5abdb9e5dc : SUCCESS in 8m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ad4b0f26274745ec98c84dca9e198899 : SUCCESS in 14m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f4ae8010c7f240b6a3fcd82cfbb4e8b0 : SUCCESS in 16m 59s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/aa63cb7e1b4e4c528366dc22c7c057d0 : FAILURE in 6m 13s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f83c1e1b2a8849b29bfb110da30bf5d6 : SUCCESS in 5m 21s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5f0b35ce90e455dbc149ed4a179d24f : FAILURE in 34m 00s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_scrubber.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_scrubber.rst has only 8 lines\u0027\n  Comments left for invalid file doc/source/configuration/glance_api.rst\n  Comments left for invalid file doc/source/configuration/glance_scrubber.rst\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file doc/source/configuration/glance_cache.rst\n  Comments left for invalid file glance/registry/api/__init__.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":1},{"id":"6e7493074e27076a0cffbef96f375601430ab0c3","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-04-06 13:11:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25c7669ea42d9b6bce0f9cb4656b1ee59811e381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:26:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb573f908b8c45cabe1d216a5ae722ea : FAILURE in 31m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c974ffbf10f74dff86c50c858b551d95 : SUCCESS in 57m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/91985b06e5a6473695cc12773cc48b4c : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6bc6e33334147e2ad9bf259c006ce31 : SUCCESS in 1h 23m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d164d30e07924524a2608bd1cb3fcd9e : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d427365f6004dad945b258371a46020 : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbcf45ca47364e06a0490bba9d5389ad : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/259609833b1941309ae1285b0b387a5c : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aef424007e27447c99816b8603aad093 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c19456e7111453885a3b16feb48a602 : SUCCESS in 7m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b566267ebaa54a0cb7ef3f08c1fa7108 : SUCCESS in 12m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/439668550aad4896ab604ef884b9815c : SUCCESS in 12m 40s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/c2d73cfbcada4011a3c28401d85491a6 : FAILURE in 7m 12s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/381079841ccd409a8f4ab5583625345e : SUCCESS in 6m 47s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b5e6c51a0c740c597a035cf7493ebb2 : FAILURE in 30m 58s (non-voting)\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_api.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_scrubber.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_scrubber.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: file doc/source/configuration/glance_cache.rst has only 8 lines\u0027\n  Comments left for invalid file doc/source/configuration/glance_api.rst\n  Comments left for invalid file doc/source/configuration/glance_scrubber.rst\n  Comments left for invalid file glance/tests/unit/v2/test_tasks_resource.py\n  Comments left for invalid file doc/source/configuration/glance_cache.rst\n  Comments left for invalid file glance/registry/api/__init__.py\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":1},{"id":"5f28b5604db6f7cc86174ca1e1ede9e136d5f3f7","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-04-23 23:34:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"feb53fc8fd461737e5dfe8d9be4f91cb434db46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 01:13:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ab1eaf8d0814dd6addf5d954ab39af8 : SUCCESS in 1h 32m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7db64caba34b41eda175fbc971f3d2bb : SUCCESS in 1h 02m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b5a92841ec32461f8df8e4ba28bd8f87 : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2ee1e76ca0a475ca148ec36c25bdfb8 : SUCCESS in 1h 26m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/577875670b064367b99fd82574107ec8 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf783342c1e467fae79bb171a28da76 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19743dbad2a84f3981dc54d430fe64cc : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2537c8213bad406e8eae7172abcb5c48 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f41bc7d431f24fe18ce78afc5436d7c8 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10dfb2e228de4f5c807f42920048f45a : SUCCESS in 6m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fe2083699bfa4f7c92f31c4c623d8202 : SUCCESS in 9m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/06764ef32fdd43df9d7676721dd31282 : SUCCESS in 17m 37s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/79b0864ffab749beb4d9fa059f6b7947 : FAILURE in 4m 32s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f88ad20644a46a19eb023f612adbaba : SUCCESS in 4m 42s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9cc1c6518154406e8d559196655e5819 : SUCCESS in 1h 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa3aff56060bb230c876cfa1eef6f3d797ab9ff5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 14:30:12.000000000","message":"Abandoned\n\nThis does not fit with our backport criteria [0]\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bed540dcdbb8251c6aa4ae0887e668cfc72f2ac","revisions":{"6bed540dcdbb8251c6aa4ae0887e668cfc72f2ac":{"kind":"REWORK","_number":1,"created":"2020-04-06 11:34:52.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/16/717716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/717716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/717716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/717716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/717716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/717716/1"}}},"commit":{"parents":[{"commit":"3e16bc7d40922f4e6176486b5e7af42989a50774","subject":"Merge \"fix properties\u0027 missing underline for VirtCPUTopology\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e16bc7d40922f4e6176486b5e7af42989a50774"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-04-08 12:49:16.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2020-04-06 11:34:52.000000000","tz":0},"subject":"Use application_url in API version document","message":"Use application_url in API version document\n\nThis handles the possible vhost in Glance API path correctly\n(like https://cloud.example.com/image) contrary to host_url.\n\nChange-Id: If86d805ddc46944e3aa2785aa5d775fe15b8468c\nCloses-Bug: #1823703\n(cherry picked from commit 33626369e591b03eb75c44b5c45c7969c7cb40e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bed540dcdbb8251c6aa4ae0887e668cfc72f2ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bed540dcdbb8251c6aa4ae0887e668cfc72f2ac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
