)]}'
{"id":"openstack%2Fglance~717889","triplet_id":"openstack%2Fglance~master~I879f4440a14b1e8420e230de84bfba5e0419a4d4","project":"openstack/glance","branch":"master","topic":"bug/1856578","hashtags":[],"change_id":"I879f4440a14b1e8420e230de84bfba5e0419a4d4","subject":"Add warning and note on image schema customization docs","status":"MERGED","created":"2020-04-07 01:21:03.000000000","updated":"2020-04-15 19:55:34.000000000","submitted":"2020-04-15 19:53:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717889-1586980382312-ef3b6f3c","meta_rev_id":"02993c45a234f8094cbd5f76a79a0dbbc775a9e1","_number":717889,"virtual_id_number":717889,"owner":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-04-15 16:17:38.000000000","_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 19:53:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-04-15 17:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-04-15 17:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 01:21:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 06:05:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-04-14 10:21:19.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-04-15 17:10:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-04-15 19:53:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3872cf4f3c837ee5a1db5e9f30d60c504d7af94","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 01:21:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7be54d67f489a1fc9f2083d11a95202d4f0f5f2e","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 01:23:59.000000000","message":"Patch Set 1:\n\ntox build failed locally with output:\n\ntodoc: failed to import module \u0027checks\u0027 from module \u0027glance.hacking\u0027; the following exception was raised:\nNo module named \u0027hacking\u0027\nERROR: InvocationError for command /home/microwave/openstack/glance/.tox/docs/bin/sphinx-build -W -b html doc/source doc/build/html (exited with code 2)\n\nThis is unexpectedly surprising since I just successfully build https://review.opendev.org/#/c/717873/2 a few minutes ago (!?). So I give Zuul a try here and see.","accounts_in_message":[],"_revision_number":1},{"id":"679d38493f155705ec2197156c1430ebdc869e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 01:26:59.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848df2165b5149c78c7b6c9f411cba0a : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654514c056f245b9a063a3abd4846ec2 : FAILURE in 4m 50s\n\nWarning:\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":1},{"id":"5c9a615921dabba48b2a9a57ad65fe16c04099e2","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 05:18:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5449c2f1ce72800ac5c7bda2c160182d8f36bed","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 05:26:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3e2f4d458fad91059bff433baaffa10f6ddc470","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 05:27:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2523105371d774748f28ee724754f7bb58f40002","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 05:30:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1e7fe8c40c094d7ae58d364e13fdce4119474873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 05:37:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647a5d666cf9462bbd98502c65b59aff : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2aac3f328a4696bd5a997d41154dd0 : SUCCESS in 6m 24s\n\nWarning:\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":5},{"id":"205383d4a620e0c3ee73b64454fd4ce7a1e1c757","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-07 06:23:28.000000000","message":"Patch Set 5:\n\nPreview of this patch is all on this page\n\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_cf2/717889/5/check/openstack-tox-docs/cf2aac3/docs/user/common-image-properties.html","accounts_in_message":[],"_revision_number":5},{"id":"bbea1cde2a458413015b81c488dbed73ef497d1a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-14 10:21:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLooks great, thanks! As this is docs only, I\u0027ll ninja it in.","accounts_in_message":[],"_revision_number":5},{"id":"4047a05e081378a9f42fe51145db98145e536aab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 14:45:45.000000000","message":"Patch Set 5: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"49cb1a3a128ebfc3e86dbdbc009f68d1a108a26c","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-15 16:17:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"459f7ff9787a05dd0fece4de5006721f558c560a","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-15 16:17:38.000000000","message":"Patch Set 6:\n\nrebase ^","accounts_in_message":[],"_revision_number":6},{"id":"e273fc265bc018a6d0b732cef6d3b8c6c13271d2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-15 17:10:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9e1a12adfa5eb2a185b4ad0f17aa484975d1f1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 17:34:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/693a1418095347d19ed103f6dfbb24f2 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bee75e792474b09bd318539d1fb6caa : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":6},{"id":"7aa5f68cf00fce5368452979f890d432eba161d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 17:34:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c57e119e3eb844939fa75cd7248ada9b9a3f8fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 19:53:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"e6301f1a86832e793234aa227a5840b5d9c4e74a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 19:53:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196e88736f4d4f17bd54927a21738c3e : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03854a9691e840b2a9822b24d03b1f35 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":6},{"id":"bd47de758cb9aa6da3432a70a69e020857a46b85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 19:55:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a9d10238b9041cfac429c740e4eb7b0 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619","revisions":{"11bb5e25e661425c05410b5d0a60e431d246694d":{"kind":"REWORK","_number":1,"created":"2020-04-07 01:21:03.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/89/717889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/717889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/717889/1"}}},"commit":{"parents":[{"commit":"d021dcc88dd02a33ad43d2e7df49ce7093ceb292","subject":"Update uWSGI doc definition names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d021dcc88dd02a33ad43d2e7df49ce7093ceb292"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 00:59:13.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 01:15:48.000000000","tz":-420},"subject":"Add recommendations on image schema customization docs","message":"Add recommendations on image schema customization docs\n\nAn operator can modify /etc/schema-image.json to include arbitrary\nproperties (which was that file\u0027s original purpose) and assign them JSON\ntypes other than \u0027string\u0027. The type is enforced by image create/update\nbut an end-user making a call that sets a value on one of these gets a\n500.\n\nThis patch add the following recommendations to the documentation:\n- Value of `type` of each property in the JSON scheme must be `string`\n- Do not delete items from the default schema-image.json file\n- If operator\u0027s need is more complicated, suggest them using metadefs\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I879f4440a14b1e8420e230de84bfba5e0419a4d4\nCloses-bug: #1856578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11bb5e25e661425c05410b5d0a60e431d246694d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11bb5e25e661425c05410b5d0a60e431d246694d"}]},"branch":"refs/heads/master"},"31c4d05fcc81d385b3f4a8ced2032f9769f54a95":{"kind":"REWORK","_number":2,"created":"2020-04-07 05:18:06.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/89/717889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/717889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/717889/2"}}},"commit":{"parents":[{"commit":"d021dcc88dd02a33ad43d2e7df49ce7093ceb292","subject":"Update uWSGI doc definition names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d021dcc88dd02a33ad43d2e7df49ce7093ceb292"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 00:59:13.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 05:17:03.000000000","tz":-420},"subject":"Add warning and note on image schema customization docs","message":"Add warning and note on image schema customization docs\n\nAn operator can modify /etc/schema-image.json to include arbitrary\nproperties (which was that file\u0027s original purpose) and assign them JSON\ntypes other than \u0027string\u0027. The type is enforced by image create/update\nbut an end-user making a call that sets a value on one of these gets a\n500.\n\nThis patch add the following recommendations to the documentation:\n- Value of `type` of each property in the JSON scheme must be `string`\n- Do not delete items from the default schema-image.json file\n- If operator\u0027s need is more complicated, suggest them using metadefs\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I879f4440a14b1e8420e230de84bfba5e0419a4d4\nCloses-bug: #1856578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31c4d05fcc81d385b3f4a8ced2032f9769f54a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31c4d05fcc81d385b3f4a8ced2032f9769f54a95"}]},"branch":"refs/heads/master"},"4704334f50882aff0dfdf17dbca8d8dff42d8893":{"kind":"REWORK","_number":3,"created":"2020-04-07 05:26:20.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/89/717889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/717889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/717889/3"}}},"commit":{"parents":[{"commit":"d021dcc88dd02a33ad43d2e7df49ce7093ceb292","subject":"Update uWSGI doc definition names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d021dcc88dd02a33ad43d2e7df49ce7093ceb292"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 00:59:13.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 05:26:13.000000000","tz":-420},"subject":"Add warning and note on image schema customization docs","message":"Add warning and note on image schema customization docs\n\nAn operator can modify /etc/schema-image.json to include arbitrary\nproperties (which was that file\u0027s original purpose) and assign them JSON\ntypes other than \u0027string\u0027. The type is enforced by image create/update\nbut an end-user making a call that sets a value on one of these gets a\n500.\n\nThis patch add the following recommendations to the documentation:\n- Value of `type` of each property in the JSON scheme must be `string`\n- Do not delete items from the default schema-image.json file\n- If operator\u0027s need is more complicated, suggest them using metadefs\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I879f4440a14b1e8420e230de84bfba5e0419a4d4\nCloses-bug: #1856578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4704334f50882aff0dfdf17dbca8d8dff42d8893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4704334f50882aff0dfdf17dbca8d8dff42d8893"}]},"branch":"refs/heads/master"},"3093a32f87106af11a9bae592a46d9b515d97d2c":{"kind":"REWORK","_number":4,"created":"2020-04-07 05:27:29.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/89/717889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/717889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/717889/4"}}},"commit":{"parents":[{"commit":"d021dcc88dd02a33ad43d2e7df49ce7093ceb292","subject":"Update uWSGI doc definition names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d021dcc88dd02a33ad43d2e7df49ce7093ceb292"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 00:59:13.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 05:27:16.000000000","tz":-420},"subject":"Add warning and note on image schema customization docs","message":"Add warning and note on image schema customization docs\n\nAn operator can modify /etc/schema-image.json to include arbitrary\nproperties (which was that file\u0027s original purpose) and assign them JSON\ntypes other than \u0027string\u0027. The type is enforced by image create/update\nbut an end-user making a call that sets a value on one of these gets a\n500.\n\nThis patch add the following recommendations to the documentation:\n- Value of `type` of each property in the JSON scheme must be `string`\n- Do not delete items from the default schema-image.json file\n- If operator\u0027s need is more complicated, suggest them using metadefs\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I879f4440a14b1e8420e230de84bfba5e0419a4d4\nCloses-bug: #1856578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3093a32f87106af11a9bae592a46d9b515d97d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3093a32f87106af11a9bae592a46d9b515d97d2c"}]},"branch":"refs/heads/master"},"9a7c66e167d32d47004aa87aececee5b412b6d4a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-07 05:30:19.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/89/717889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/717889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/717889/5"}}},"commit":{"parents":[{"commit":"2d21685ee44b8a757078dce84bca8e937f3f7da6","subject":"Merge \"Remove all references to sheepdog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d21685ee44b8a757078dce84bca8e937f3f7da6"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 00:59:13.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 05:29:34.000000000","tz":-420},"subject":"Add warning and note on image schema customization docs","message":"Add warning and note on image schema customization docs\n\nAn operator can modify /etc/schema-image.json to include arbitrary\nproperties (which was that file\u0027s original purpose) and assign them JSON\ntypes other than \u0027string\u0027. The type is enforced by image create/update\nbut an end-user making a call that sets a value on one of these gets a\n500.\n\nThis patch add the following recommendations to the documentation:\n- Value of `type` of each property in the JSON scheme must be `string`\n- Do not delete items from the default schema-image.json file\n- If operator\u0027s need is more complicated, suggest them using metadefs\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I879f4440a14b1e8420e230de84bfba5e0419a4d4\nCloses-bug: #1856578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a7c66e167d32d47004aa87aececee5b412b6d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a7c66e167d32d47004aa87aececee5b412b6d4a"}]},"branch":"refs/heads/master"},"bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619":{"kind":"REWORK","_number":6,"created":"2020-04-15 16:17:19.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/89/717889/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/717889/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/717889/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/717889/6"}}},"commit":{"parents":[{"commit":"fa18f745df3a5200637e661fd04f751b5e8bb595","subject":"Merge \"Revise admin interoperable image import docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa18f745df3a5200637e661fd04f751b5e8bb595"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-07 00:59:13.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-15 16:16:30.000000000","tz":-420},"subject":"Add warning and note on image schema customization docs","message":"Add warning and note on image schema customization docs\n\nAn operator can modify /etc/schema-image.json to include arbitrary\nproperties (which was that file\u0027s original purpose) and assign them JSON\ntypes other than \u0027string\u0027. The type is enforced by image create/update\nbut an end-user making a call that sets a value on one of these gets a\n500.\n\nThis patch add the following recommendations to the documentation:\n- Value of `type` of each property in the JSON scheme must be `string`\n- Do not delete items from the default schema-image.json file\n- If operator\u0027s need is more complicated, suggest them using metadefs\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I879f4440a14b1e8420e230de84bfba5e0419a4d4\nCloses-bug: #1856578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
