)]}'
{"id":"openstack%2Fglance~718287","triplet_id":"openstack%2Fglance~master~I840f3cbeda28da8b02dd141fde582c9110aeb21e","project":"openstack/glance","branch":"master","topic":"bug/1870336","hashtags":[],"change_id":"I840f3cbeda28da8b02dd141fde582c9110aeb21e","subject":"Update \u0027common image properties\u0027 doc","status":"MERGED","created":"2020-04-08 01:01:03.000000000","updated":"2020-04-14 15:02:39.000000000","submitted":"2020-04-14 14:45:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718287-1586875544489-69063b05","meta_rev_id":"188a9b12818d5234d5927d820fcc06ee55feab03","_number":718287,"virtual_id_number":718287,"owner":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 14:45:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-04-14 10:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-04-14 10:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-08 01:01:35.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 01:01:40.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 01:01:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 01:01:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-04-14 10:18:47.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-04-14 14:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49e1bf4cf6d7651d62224e680e2909e88eb161d4","author":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"date":"2020-04-08 01:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"833ee4f2354600128500ce26e181b9e5f4eee62d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 01:36:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7592d934cb2d4f85b3103adf167bf0e7 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2798e1d662c04ba6978337b2da02c7a0 : SUCCESS in 7m 36s\n\nWarning:\n  Comments left for invalid file glance/cmd/api.py","accounts_in_message":[],"_revision_number":1},{"id":"bcbfbf534f47c45ac99356168e242a7f395ca332","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-04-14 10:18:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good, as this is docs only change, I\u0027ll ninja it in.","accounts_in_message":[],"_revision_number":1},{"id":"01b055a777cebc276335e0460a669807cdaddb3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 10:33:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc6490b9a3bf0d665899f2434b7a175a589c01e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 14:45:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316713057d2a49c683ff024caa9245b1 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4931fe7d03f2460d88d55bb5d3c4de78 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a252df337c1c8fced033efade477652a36df6ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 14:45:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"febca670ed22e27a4feff143923a7994faa3b5fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:02:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71072eaa7a2342ec9731a15a707dea45 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"984e844c6ab3b365f5104e36429bda0ebd5e9a9d","revisions":{"984e844c6ab3b365f5104e36429bda0ebd5e9a9d":{"kind":"REWORK","_number":1,"created":"2020-04-08 01:01:03.000000000","uploader":{"_account_id":31209,"name":"kayluu","email":"freefood@osuosl.org","username":"khashf"},"ref":"refs/changes/87/718287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/718287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/718287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/718287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/718287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/718287/1"}}},"commit":{"parents":[{"commit":"2f2e118a58e5ef2e2b91036bc7069b6a6910a471","subject":"Merge \"Add Policy enforcement for several Metadata Definition delete APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f2e118a58e5ef2e2b91036bc7069b6a6910a471"}]}],"author":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-08 00:44:39.000000000","tz":-420},"committer":{"name":"khashf","email":"freefood@osuosl.org","date":"2020-04-08 00:44:39.000000000","tz":-420},"subject":"Update \u0027common image properties\u0027 doc","message":"Update \u0027common image properties\u0027 doc\n\nUpdate doc/source/user/common-image-properties.rst (live link [1]) to be\nin sync with etc/schema-image.json\n\nThis patch does 2 actions:\n1. Add the missing properties that are in etc/schema-image.json to\n   the doc\n2. Rearrange the order of appearance of properties in the doc to be\n   the same as they appear in etc/schema-image.json\n\n[1] docs.openstack.org/glance/latest/user/common-image-properties.html\n\nChange-Id: I840f3cbeda28da8b02dd141fde582c9110aeb21e\nCloses-bug: #1870336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/984e844c6ab3b365f5104e36429bda0ebd5e9a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/984e844c6ab3b365f5104e36429bda0ebd5e9a9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
